)]}'
{"id":"openstack%2Foctavia~791650","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I602418264e171a2b1a926eff0b1f9e6dc186295a","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I602418264e171a2b1a926eff0b1f9e6dc186295a","subject":"Validate user access to vip_subnet_id when creating a LB","status":"MERGED","created":"2021-05-17 07:12:38.000000000","updated":"2021-05-19 20:36:18.000000000","submitted":"2021-05-19 20:34:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791650","meta_rev_id":"fc96b88e3d4f8b66008b9022bc181e5b138aceb4","_number":791650,"virtual_id_number":791650,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-19 20:34:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2021-05-19 17:46:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2021-05-19 17:55:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2021-05-19 17:55:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 07:12:38.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-05-17 09:09:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-19 17:46:09.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"b7d87fd6da82e1a9a1115c45dfbdfba333edacf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-17 07:12:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20bbc185ad0df73e4951a668c87ea83a97923700","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 09:09:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0aaa9fc130b64f39bc436047fe6963a3 : SUCCESS in 25m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ddb4551f92c4eb79a44ad0492226419 : SUCCESS in 18m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7456d6e9b01a43ca95d65ecff1f3afbf : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8abc41da1bc64acab9d4b7cdd70b9877 : SUCCESS in 19m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d0560f818aa4c6f964e75467b8e2550 : SUCCESS in 16m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c0dbe2f945c450cbd180a77bb30dac8 : SUCCESS in 12m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/270946c81a824110b9ec3d37e4fcf45e : SUCCESS in 5m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1fe71de8dc634e2ca508b0c0a3a1587e : SUCCESS in 18m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1e4e7ab019d244bd9b0811073bec4091 : SUCCESS in 20m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e885fbc239be47f69737a66d45a1dffa : SUCCESS in 7m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c83bd524911f442c87ed81c0bfcc892b : SUCCESS in 18m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/40843a7f784842a0a46b9a95c1251589 : SUCCESS in 1h 45m 39s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ee98918abedb4f30981fe6886c163b0a : SUCCESS in 1h 38m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/247041ac671b408d973d4918e2ee9a34 : SUCCESS in 1h 33m 38s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6b779df490a3408e90dc9c61790e1374 : SUCCESS in 1h 46m 17s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e5443570642045d59ab41d23a3cb21a5 : FAILURE in 15m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3173acb35fdb4520b2294e5c0c9b6791 : FAILURE in 58m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/34581adb175d488d8567c02111e6c6a2 : FAILURE in 46m 25s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3291e9f9953d4e2da871221532e9fb11 : SUCCESS in 37m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/939394343e9a4f5592a030a6630bffc7 : SUCCESS in 1h 54m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/92975f03d0eb43609e1d2247c43560d3 : SUCCESS in 40m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/646764e013554a04bffd7eff5a9eb72a : SUCCESS in 43m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/452498511e564f2699f12d10ec13d528 : FAILURE in 18m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3dbd9583292246381a9a80623097ac6f25c3efe","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-19 17:46:09.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"ece179d230988eee432533c5c70b9c91cf5cf4b9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-19 17:55:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b1dbde714c95dc502af2368d60341adb48c18ecc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 17:55:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8be0c15d94c69b77c47c106f88d1b1f1c1bf3f58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 20:34:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84ee6e39173d45249b1118fc9f7629b6 : SUCCESS in 19m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7d0743005c11498f9b3572c9d6119a26 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/536b0cfd0e93489d875bdd2f8311f44d : SUCCESS in 19m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d1065da1d5ca4d4e9b9477edc890b9d7 : SUCCESS in 18m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b16a4c94eb94024aa298140c845066b : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0aa2ccd4d994a94ad071887229d8576 : SUCCESS in 5m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/cee042c1cfe74be6a8742273f72e84f2 : SUCCESS in 7m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bae21b0171584b6590f8c46f27c24e6b : SUCCESS in 19m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/91fdda953c83436e8c8f040100214ee0 : SUCCESS in 1h 48m 12s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/6c0f4836d3ae470b853099a5425017b8 : SUCCESS in 2h 04m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd0fa2b3ef7246f091c5d645ff68a1c7 : SUCCESS in 1h 56m 14s","accounts_in_message":[],"_revision_number":1},{"id":"bd810fe64e58539f3d9694a79c91f45fd9b6d1b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 20:34:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fc96b88e3d4f8b66008b9022bc181e5b138aceb4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 20:36:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e830ab5d594b4379acc00b3636a532fa : SUCCESS in 1m 34s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc1a8cf19e374a4dba0b0de28a1f46ad : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"95972fc1129dacab5811f04b62fb06c5c2cec639","revisions":{"95972fc1129dacab5811f04b62fb06c5c2cec639":{"kind":"REWORK","_number":1,"created":"2021-05-17 07:12:38.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/50/791650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/50/791650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/791650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/791650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/791650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/50/791650/1"}}},"commit":{"parents":[{"commit":"db24d7bb5187676c0501e722b7eaa5bd9161829e","subject":"Fix weighted round-robin about UDP listener","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db24d7bb5187676c0501e722b7eaa5bd9161829e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-02 18:58:14.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-05-17 07:12:22.000000000","tz":120},"subject":"Validate user access to vip_subnet_id when creating a LB","message":"Validate user access to vip_subnet_id when creating a LB\n\nA user was able to create a LB using a vip_subnet_id from another user\n(by passing the UUID).\nNow, the vip_subnet_id parameter is validated using the user context,\nso an error is returned if the subnet doesn\u0027t belong to the user.\n\nI479019a911b5a1acfc1951d1cbbc2a351089cb4d was a previous attempt to fix\nthat bug but vip_subnet_id check was missing.\n\nStory: 2008586\nTask: 41741\n\nDepends-On: https://review.opendev.org/774157\n\nChange-Id: I602418264e171a2b1a926eff0b1f9e6dc186295a\n(cherry picked from commit 8d86187c0abd282d10a9fbe294df68f9100ad4a0)\n(cherry picked from commit 7d1b81d78f4383af725e0f6557b6dfe37ec74799)\n(cherry picked from commit 73db7b076290aca89544f389decd3b85c14fad0c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95972fc1129dacab5811f04b62fb06c5c2cec639"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95972fc1129dacab5811f04b62fb06c5c2cec639"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
