)]}'
{"id":"openstack%2Foctavia~791652","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I987535e90cbf13917c50c5905c64b614188928ec","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"I987535e90cbf13917c50c5905c64b614188928ec","subject":"Fix devstack cleanup when using amphorav2","status":"MERGED","created":"2021-05-17 07:14:08.000000000","updated":"2021-06-05 00:45:19.000000000","submitted":"2021-06-05 00:42:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791652","meta_rev_id":"e00c07d8d4747afe01060a0cf8729f65be48b588","_number":791652,"virtual_id_number":791652,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-05 00:42:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-05-31 07:54:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-05-18 15:15:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-06-04 22:30:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2021-06-04 22:30:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 07:14:08.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-05-17 09:26:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-18 15:15:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-05-31 07:54:36.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-06-04 22:30:07.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"00d3d528911810f29a688cd662d29dd1a2409b9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-17 07:14:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76e16990c369daf07dfc06c6abe86e841e2b32c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 09:26:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4818fe6e22f3411eb1d7413fd2306e18 : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3894300c165d4a4a860ac2d263d07c2c : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5ea5141b4d74afab8c18c607de99808 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1059abca464d4776a613ed7aa39e4b4c : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bc5ee339cf64f18aca5f1e5220507b8 : SUCCESS in 18m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cdf949758e964eb7a8c56c3827e20a58 : SUCCESS in 7m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5fbf47133a914f8baea060e190abb421 : SUCCESS in 8m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0e51dd8444f94ce48d2465a2bd9856b5 : SUCCESS in 7m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1a0cb4cd9f86482dbc1a70efec34a330 : SUCCESS in 9m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6a76a797fc384ca9b619710735a32d13 : SUCCESS in 1h 39m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c3a24c9d6fe14d3db3fcb21d6abbe96b : SUCCESS in 1h 45m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e3fab4749fa146edaed73f152d45f5a0 : SUCCESS in 43m 02s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3df1de25252542c2b05d4a3db00be3ed : SUCCESS in 42m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68ac9b51844e4c539f120ea2ff0b33e7 : SUCCESS in 43m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/42c7da49c7a547fa84a66d7564966ed7 : SUCCESS in 1h 03m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1a2ba2a6075f4bcd8653c272221efd92 : SUCCESS in 2h 09m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1ee0a9f49ba141dca78ac27ae8176643 : FAILURE in 1h 51m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e0413521c6b74512917a10641073a4cf : SUCCESS in 1h 27m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f3231ebaf8e246abb0838c34406251d3 : SUCCESS in 1h 50m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9d29f95eeee84286afb4575e40ce264c : FAILURE in 1h 28m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/219a6b3f9bc54c5f864f4262219b817b : FAILURE in 7m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59910b7a6a8bb2d6004ddb668516f6628458cc2d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-18 15:15:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f3465d7cf6d71cc6f09c07ff97878297f1a0c6ba","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-05-31 07:54:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"79c5fb3fe824f6c3f8458f24b90a8cacaa9a7a1f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-04 22:30:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"7c4b2c81663d462d5fcb4e91d44b37baed666dbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 22:30:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95fef322babdbef25c6c298a1748107726f02030","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 00:42:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c74f15b55cfd47beb33112ae224080a0 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fcf264f5b734fbf86e7fdccff58b876 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15e26209e7d44555ac1c469b45716dd9 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/105d29235c944ee1af60832dec3d1d81 : SUCCESS in 11m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f5b6827645a2499d81c2d0386b4299b0 : SUCCESS in 7m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c4c616d49844f458ed0cd0dda96cdb3 : SUCCESS in 7m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3316ce8a375a4e94a8a474de0c191426 : SUCCESS in 1h 58m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a195cf9ce45e43599b6eb7916b18d01b : SUCCESS in 2h 03m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/97c43993338b409d8e572d1ae0221be5 : SUCCESS in 34m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6ecfe13da7e647079593f3a66dc6aac6 : SUCCESS in 1h 01m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6a7e6ec9268a4f8f956a7b608a4945c2 : SUCCESS in 45m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3688871d45e94f25b5b7152094ae1a15 : SUCCESS in 56m 31s","accounts_in_message":[],"_revision_number":1},{"id":"a0bcbf3e17c555f7b0dad9e3f7818c24a87fadd3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 00:42:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e00c07d8d4747afe01060a0cf8729f65be48b588","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 00:45:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef762537085b4e7b8bfe447a78ccabc8 : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d49484eb6ae4fc1cf59c110a725d33aaad0b038","revisions":{"5d49484eb6ae4fc1cf59c110a725d33aaad0b038":{"kind":"REWORK","_number":1,"created":"2021-05-17 07:14:08.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/52/791652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/52/791652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/791652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/791652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/791652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/52/791652/1"}}},"commit":{"parents":[{"commit":"035d6d6daa86488a3a1da8a081a65ba9d21436c9","subject":"Merge \"Fix empty Batch Member Update to unlock objects\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/035d6d6daa86488a3a1da8a081a65ba9d21436c9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-25 06:45:35.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-05-17 07:13:57.000000000","tz":120},"subject":"Fix devstack cleanup when using amphorav2","message":"Fix devstack cleanup when using amphorav2\n\nWhen cleaning up a devstack environment with amphorav2 enabled,\n./clean.sh might fail because redis has already been stopped and\nuninstalled.\nThis commit move the stop_redis step to the octavia_stop step and\nensures that stopping redis doesn\u0027t fail.\n\nChange-Id: I987535e90cbf13917c50c5905c64b614188928ec\n(cherry picked from commit 5d470357ec6acd7f31a626283be14154968c0798)\n(cherry picked from commit 168f1a121266bd0c4ae15b1887f8e9e14022824d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d49484eb6ae4fc1cf59c110a725d33aaad0b038"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d49484eb6ae4fc1cf59c110a725d33aaad0b038"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
