)]}'
{"id":"openstack%2Foctavia~791653","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I987535e90cbf13917c50c5905c64b614188928ec","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I987535e90cbf13917c50c5905c64b614188928ec","subject":"Fix devstack cleanup when using amphorav2","status":"MERGED","created":"2021-05-17 07:16:53.000000000","updated":"2021-06-17 22:25:14.000000000","submitted":"2021-06-17 22:23:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"791653","meta_rev_id":"cfa62c55ccb8ef9e333256154e5fd9fe82512d83","_number":791653,"virtual_id_number":791653,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 22:23:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-06-17 22:25:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-06-17 22:25:14.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-17 22:25:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-17 20:36:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 07:16:53.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-05-17 09:19:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-18 15:16:49.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-05-31 07:54:05.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-06-04 22:32:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"1bfe13c10354d7d7cec4e3a76a38a02f2c6a2f75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-17 07:16:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8c6e09880dd8036686b61efa38d75b0ecbf435e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 09:19:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d6f65fbebdf45b9aed417229810f5b1 : SUCCESS in 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6767049343a14002adb14dbb6664d764 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52d6ab8b59a84c71b5684557cd38b2e4 : SUCCESS in 7m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5164dfbc89cf47e39e446ac8abc1801e : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce4a1faaecf14157b037c4cb9f4e89d1 : SUCCESS in 18m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/795459cd9a6f49c394977afa8823f7d5 : SUCCESS in 9m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/aa4c57a16d9645e2a3317bebbfe56b96 : SUCCESS in 9m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9e18aded77fe4dcab506e24b6c04b22a : SUCCESS in 9m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c83da8ca80384c5696d9ecf7b1630d0d : SUCCESS in 1h 29m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1f4cb589999741609bbf340aabf6c331 : SUCCESS in 1h 44m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a8b05a44b56841f4b5f48651916d65c5 : SUCCESS in 56m 25s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9d17372894f2420a9221915a270eab41 : SUCCESS in 40m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/65ee1b4ad09c43bd92910cb9bc4598c5 : SUCCESS in 37m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/95378a85c94646f5b1906ccbf133f748 : SUCCESS in 1h 28m 14s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0fd4f5e7b75d4151ab0d599a1e16a58d : SUCCESS in 1h 52m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/62b276837b6048189319dd957d1a56f6 : FAILURE in 50m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/92299f41db894c57836746446a55a073 : SUCCESS in 1h 59m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1bce0a1d303c4e006b0b2495430ed0119dc029a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-18 15:16:49.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5681d8c75b6e59e71c31db066057cd22cbc02168","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-18 15:23:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"188b5d5ed535c31b793104a73d4d0b2ce779ba6f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-18 15:23:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"969ce5ca75ac92e614bc058116f03cee549b1fff","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-18 15:25:53.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"41e74f71dab84ac71851c2aec8d5c5bcbc8029fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 17:10:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/946e7f6b8e4d405a9c3f7476b96d85da : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9fae7033a0c4ee2a59b6ff68c3cd012 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e36386bc16a4930890bd4d0dd51e03d : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db73e3a6cf51457eb407de317fce245a : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed698ecda9946cfb6faa86d661c4953 : SUCCESS in 10m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3bfb6ff695e945ac8afdfdaea421cb06 : SUCCESS in 6m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1ec46b195a0e41ecac8209e37557e508 : SUCCESS in 10m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8f94f1de98384062a772d1111ff11110 : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a6aa36daba2d4b53b33b1aafaec9d995 : SUCCESS in 1h 03m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5993a1fa73554503875726416e0dce01 : FAILURE in 1h 30m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b83a3d25045d49c18ad48ac5b2fb74a4 : SUCCESS in 38m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b13249741fb74a1db5eea3da52fea3c9 : SUCCESS in 34m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/781e543fa8ef49de967e861dc0b4e346 : SUCCESS in 33m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/168c8130defc410e9bc22177d88810a1 : SUCCESS in 1h 07m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3216666cc8034d54a530831dab1a23bd : SUCCESS in 1h 45m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/862a16a7ffe2466584b13a71ea99a1d7 : FAILURE in 43m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c1ec2e23440f49c29b1542921eddec14 : SUCCESS in 1h 35m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c2627a982bda368e2a8af85687df041cba7989ad","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-18 17:13:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"16ed62e8b3b8175ef3feda066322117c56235cde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 19:37:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ac348fb4c274953afb6ecc1d840da4c : SUCCESS in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/469956dd48ae4e9c99bbd3f6111f37d6 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1e5344893754fafb7c9adecdfd67e14 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd95aec2883746ab8a46868a0eac2f59 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69a1f35349784dc78c76bc6c8142c3bc : SUCCESS in 11m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cf6e11d46fb142549203bbe17681b4bf : SUCCESS in 5m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/00dad7512c5446aebad322290519f63b : SUCCESS in 8m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/60291d9e7803436988c125a80eb316cc : SUCCESS in 7m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a3758a59935b47029975603e5dd35403 : FAILURE in 1h 57m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c7d3eb7e1bcf4d9a84436180e87ab710 : SUCCESS in 1h 41m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b9daf93b0b08446b8e17d839ab54b2da : SUCCESS in 58m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6983eb1ffc3c41b68c91385b99e4aa2a : SUCCESS in 37m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/591984c6493a4a8f9a0960eba9accc3f : SUCCESS in 44m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4eb033a674cd468bbaf74eab752f5628 : SUCCESS in 1h 33m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3a88b7ca7b364471abd8ff0cd37f0133 : SUCCESS in 2h 21m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/efdf242a6d19423fa12d2c1bb989c574 : FAILURE in 1h 01m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/013eb9ff89db4af3ae38fadbe68d746d : SUCCESS in 1h 59m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eadd004b775476ddb318a7174cc53db560e8b445","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-18 20:11:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4265796fbecb713a84e70bd11de058113ae44037","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 22:06:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce0d8fed05474ecfb0935118e2c1b8c5 : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9e8df1ceda643c09fecc029798712e6 : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ff926ccad4e4fb8864cf1f337cc3886 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e5bafc62fb83445fbab402fcf1720eb7 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/960d2c1bd7f849a78f77bdfcca155fe7 : SUCCESS in 18m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c22b40a908ae4c8db7e80bc0edf0e385 : SUCCESS in 6m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5e85ec70466047d8a55f71671fe8d688 : SUCCESS in 7m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/acaa266cf303410796a5aedcf12d0f10 : SUCCESS in 7m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2ac32d0a3cd74d7db1aecf89a50a714b : SUCCESS in 1h 17m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0734ab829b7544979f88cbe4f4313ba5 : SUCCESS in 1h 49m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/025b3dcd3c6a4b49837ad995519c8c3c : SUCCESS in 37m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a91bd6844ccd4f8c8f7e77780f6dcb66 : SUCCESS in 32m 53s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f8e68008a804d3ba3b9db4eecb383f0 : SUCCESS in 39m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7bf3305e8fc648e7b95fbf50bf158172 : SUCCESS in 1h 02m 39s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2b0178af5e6c40d686aca25756cbf642 : SUCCESS in 1h 27m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/65b19cf00fed47b78446ebafe90306c8 : FAILURE in 1h 39m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bf969d355e2342eeb4be318f64135671 : SUCCESS in 1h 53m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f431337adb0e4b512beff5a9bbbf581fd09e87b9","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-05-31 07:54:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f02f17ad745d05609f3bff63d670d0c179bbe8ea","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-04 22:32:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"69dedc17957fd0244c10678bab418f5cfd480018","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 22:33:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7e7b0834660395dd0073f80b6895ccda4a1692d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 00:43:01.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/532d8abb4f864655829ccc2cce473127 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfb37b2b008e47578e9cfea282ae821c : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0aa7ac2a0eb64004a317d2a8bde7252f : CANCELED in 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07ff368cdb5e4ce7bbec378980a062a9 : CANCELED in 6m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/db79a04e327f48a4b8ccf511fc571dd8 : SUCCESS in 7m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fbceb3084b6e4c959ed036aa5e8798fd : CANCELED in 6m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c2c5aa6cde074ade84a06443e6c0b092 : FAILURE in 6m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6a520433abb34275bb8ffdb20ad3cead : CANCELED in 6m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d98479723350454e80305624ea33b84a : CANCELED in 6m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd88c2a99e914760aae757378f7f98f5 : CANCELED in 6m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cc6ec08723da4120bb30ed85afb4345c : CANCELED in 7m 45s","accounts_in_message":[],"_revision_number":2},{"id":"5dd22ffae59e715a40c92d0c4db5b32f185e1173","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-08 12:15:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"48fb9413b2e97c62002c7eba3448f3c9f5dbf293","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 13:50:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6cbf5427d774ea0ad3cac64b3c6300d : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cda31e26430426daf4c5e585fcbf52f : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16653f1b7057481e938e8a9ac1fb07a7 : SUCCESS in 7m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f478de158735439989e6454a8d8cb28d : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93ca42ea12a74e9690320424f7f61872 : SUCCESS in 13m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6f154e43fc714cd188035026aea05252 : SUCCESS in 7m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/81243076750147b7aa640548060a7f93 : SUCCESS in 8m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f2ca3cee9af4491aa25773953d97eaa7 : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/55a48d91e2b54985830d9a493d64d3af : SUCCESS in 1h 32m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d95475f10e8044f796248148e30a9360 : FAILURE in 9m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ec59c2080d30469eab8ba3da8e6449a6 : FAILURE in 7m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/aa10703daeb84525813a4e5d64194910 : FAILURE in 8m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d76ded4993b463fb2a14899c4012a39 : FAILURE in 8m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5a6ec1c517854c65a37efbf6fd796065 : SUCCESS in 1h 20m 40s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dde7e9f069eb45adac52d19c79932ebb : FAILURE in 9m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e88eb8a8c0d147c1ab2857156de123c2 : FAILURE in 12m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d5ac70b85ffe46c7a2a3811dbe31175b : FAILURE in 8m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"df207709e5362a2a08ae86bf6faa61e6d00b28f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-06-08 14:48:41.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"5c57144dd98b3d6855bee24ea94ae6f5c60f8569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 17:06:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce7fe72e125543ccb342023bf895a7f4 : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dfad310f69e41dfbbb75250c7afd945 : SUCCESS in 8m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d21b93bdc6f42a18086d037074005b4 : SUCCESS in 7m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a497aac2f874cb7a52f0e7d8656f999 : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0216345fdf9144acad0203e761ba7736 : SUCCESS in 14m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0d0c4c7483dd4fa0a0ec51a4aefae02d : SUCCESS in 7m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/893919806e7841bda065d79c5ff2bed3 : SUCCESS in 10m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b6b3769057664bc18afcbf6b65a65f34 : SUCCESS in 8m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2966c39267eb49f1a20d636bf54d4aeb : SUCCESS in 1h 40m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/05d39b88270641e0a9df6022ae6cb40f : SUCCESS in 1h 59m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6523253957b945faa899677b9bf1aa96 : SUCCESS in 45m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/55e5503e756144c38558cd1204cf20af : SUCCESS in 48m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/336382d35df44f7dab95e2b82d55ff2e : SUCCESS in 49m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cb479d77629e4912a9bf11454138f5ab : SUCCESS in 1h 13m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f1198b7f7c64461c8e9c2ff9bb58eb08 : SUCCESS in 1h 40m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/21bef60a74754b21b46fc32534526bb0 : RETRY_LIMIT in 8m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ba64389d96344d90bbcc564d99eedffd : SUCCESS in 2h 12m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7bb4dc76749e7761a1c73d0ee100dff4fee68991","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-17 20:36:33.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"85b14339dd82e8e9892c28a17e4daafd5069155d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 20:36:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"22dab1a5e5d451fef33bd92abeff263e6c3297d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 22:23:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/480481834474485f833d7019d61b0548 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c84e1c8d4974302ad2e931b41d6c260 : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a2f0a932feb40fe9b7e3354130d3abe : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2f894a828e24400ba1c6e266fbb5511 : SUCCESS in 13m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b3b7f15e0d314cfa8e2cd950b67a6685 : SUCCESS in 8m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/359183d7daa6400eb8fd51941b8fb1e4 : SUCCESS in 1h 04m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6182c9e213b44e2fbcc73c9f9bd37d83 : SUCCESS in 1h 45m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39441f877d134335a0b0e6a6623179de : SUCCESS in 36m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d4432b7e763c4c84858d3ad2000a7957 : SUCCESS in 57m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28a8253808fd491bb4d76e04ba3c80df : SUCCESS in 43m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7061af91ed0146dfbb35b52da0b7e29f : SUCCESS in 50m 54s","accounts_in_message":[],"_revision_number":3},{"id":"17d6e343d9d11e6bb4a7b915460cb2cadf15f613","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 22:23:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8b3e44af93c3b8622f2ed4e837811be85e14bc9f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 22:25:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3d03500ca144f5d85a2b624b28dd988 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8829f0fbbe6b100baba16214f8829f65cc828952","revisions":{"c42fe89f5db16b1a3058009d1423ab3bf50ea21f":{"kind":"REWORK","_number":1,"created":"2021-05-17 07:16:53.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/53/791653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/791653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/791653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/791653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/791653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/791653/1"}}},"commit":{"parents":[{"commit":"46b9ce7cde29cce0a8cf94fcb2ca8587d2a71bdf","subject":"Cap hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46b9ce7cde29cce0a8cf94fcb2ca8587d2a71bdf"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-25 06:45:35.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-05-17 07:16:35.000000000","tz":120},"subject":"Fix devstack cleanup when using amphorav2","message":"Fix devstack cleanup when using amphorav2\n\nWhen cleaning up a devstack environment with amphorav2 enabled,\n./clean.sh might fail because redis has already been stopped and\nuninstalled.\nThis commit move the stop_redis step to the octavia_stop step and\nensures that stopping redis doesn\u0027t fail.\n\nChange-Id: I987535e90cbf13917c50c5905c64b614188928ec\n(cherry picked from commit 5d470357ec6acd7f31a626283be14154968c0798)\n(cherry picked from commit 168f1a121266bd0c4ae15b1887f8e9e14022824d)\n(cherry picked from commit 5d49484eb6ae4fc1cf59c110a725d33aaad0b038)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c42fe89f5db16b1a3058009d1423ab3bf50ea21f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c42fe89f5db16b1a3058009d1423ab3bf50ea21f"}]},"branch":"refs/heads/stable/ussuri"},"a8321d1f153e4a3294b87c51969822f01fffcbcd":{"kind":"REWORK","_number":2,"created":"2021-05-18 15:23:40.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/53/791653/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/791653/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/791653/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/791653/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/791653/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/791653/2"}}},"commit":{"parents":[{"commit":"46b9ce7cde29cce0a8cf94fcb2ca8587d2a71bdf","subject":"Cap hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46b9ce7cde29cce0a8cf94fcb2ca8587d2a71bdf"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-25 06:45:35.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-05-18 15:23:26.000000000","tz":120},"subject":"Fix devstack cleanup when using amphorav2","message":"Fix devstack cleanup when using amphorav2\n\nWhen cleaning up a devstack environment with amphorav2 enabled,\n./clean.sh might fail because redis has already been stopped and\nuninstalled.\nThis commit move the stop_redis step to the octavia_stop step and\nensures that stopping redis doesn\u0027t fail.\n\nChange-Id: I987535e90cbf13917c50c5905c64b614188928ec\n(cherry picked from commit 5d470357ec6acd7f31a626283be14154968c0798)\n(cherry picked from commit 168f1a121266bd0c4ae15b1887f8e9e14022824d)\n(cherry picked from commit 5d49484eb6ae4fc1cf59c110a725d33aaad0b038)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8321d1f153e4a3294b87c51969822f01fffcbcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8321d1f153e4a3294b87c51969822f01fffcbcd"}]},"branch":"refs/heads/stable/ussuri"},"8829f0fbbe6b100baba16214f8829f65cc828952":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-06-08 14:48:41.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/53/791653/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/791653/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/791653/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/791653/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/791653/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/791653/3"}}},"commit":{"parents":[{"commit":"19619ce88862b17c4fdbf81072fa77e5252e533e","subject":"Validate user access to vip_subnet_id when creating a LB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19619ce88862b17c4fdbf81072fa77e5252e533e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-25 06:45:35.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-06-08 14:47:15.000000000","tz":120},"subject":"Fix devstack cleanup when using amphorav2","message":"Fix devstack cleanup when using amphorav2\n\nWhen cleaning up a devstack environment with amphorav2 enabled,\n./clean.sh might fail because redis has already been stopped and\nuninstalled.\nThis commit move the stop_redis step to the octavia_stop step and\nensures that stopping redis doesn\u0027t fail.\n\nChange-Id: I987535e90cbf13917c50c5905c64b614188928ec\n(cherry picked from commit 5d470357ec6acd7f31a626283be14154968c0798)\n(cherry picked from commit 168f1a121266bd0c4ae15b1887f8e9e14022824d)\n(cherry picked from commit 5d49484eb6ae4fc1cf59c110a725d33aaad0b038)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8829f0fbbe6b100baba16214f8829f65cc828952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8829f0fbbe6b100baba16214f8829f65cc828952"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
