)]}'
{"id":"openstack%2Foctavia~791657","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I3125cfc7b3bd12eed139b058fc2bb7f23f12abc7","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"I3125cfc7b3bd12eed139b058fc2bb7f23f12abc7","subject":"Fix rsyslog configuration when disabling logs","status":"MERGED","created":"2021-05-17 07:23:06.000000000","updated":"2021-06-18 12:36:50.000000000","submitted":"2021-06-18 12:33:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791657","meta_rev_id":"2b192294917223bf4f939da89a38d42560d7797c","_number":791657,"virtual_id_number":791657,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-18 12:33:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-06-08 12:18:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-06-04 22:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-06-08 12:18:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 07:23:06.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-05-17 09:50:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-04 22:34:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-06-08 12:18:46.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"2ca11628d144c8a5a2c08af5911fb84fc1575792","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-17 07:23:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c9e7f18a38a7d2d30a78db8cb12f73b3f3e69ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 09:50:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f03eec52d5647e4a95497d931233ae4 : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ece852e4769b40f1b2b2cf411b9b9bd2 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6afc1a18c191405293162ca6d71f1f04 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/647e187e3da9486186c7d42f6c858536 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc519a35c9304d4fae50ecafbd84017c : SUCCESS in 12m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca480f7e82bf43fba814477e1fc30460 : SUCCESS in 6m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d3183d62206042eeaf7b934b94739b5a : SUCCESS in 6m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/627326c5fd3742829b2bca420ed6f4ef : SUCCESS in 10m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7146de2464234c4abf68161b6d32e30c : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7d16d9d9a78a4fb69ed70a93e0bd41dd : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2aa88004d34640809b1f60cd929f809a : SUCCESS in 1h 33m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2eb05199c78343c48d2a1b26d3509792 : SUCCESS in 1h 50m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3bf8406eaecf48f08644d3646548a6ba : SUCCESS in 1h 29m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c7615ed94200486b9b103939507e51ba : SUCCESS in 47m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f69a03275596478aad25e2df0f779310 : SUCCESS in 37m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2297ca8457c347ca964eafcd26d228d6 : FAILURE in 48m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5c619ec2ddc946219aeb3894046e2f2c : FAILURE in 2h 06m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b87cde4e1e174961b45344215c73c3ca : FAILURE in 2h 05m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/39dd382b89504e57b0232d10a3744252 : SUCCESS in 1h 36m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/eb369d4d077d4085bad0b9f75c9055c5 : SUCCESS in 1h 58m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/954819cd92da43ff8ee33ae334a34f43 : SUCCESS in 1h 48m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c6d47858f3d249b09e1849afb5a6f314 : FAILURE in 8m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2105dbb79b211862432077428bd77cafc9288101","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-17 09:57:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"502c1308151f21d1909e6339123ba9039e857f6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 12:20:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8f11135858f4f8e96618ba56cdc0e82 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76e135f9e5f64dfdbeca55487ad5cea0 : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd0cd04e6da946469405b78acbb600f5 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e641da50f30b43b8a3de0753cc8d2bc1 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/517e140177be422eac4b89b33912cf2e : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d79e3dc4f68d419e9bb145b926a2e02b : SUCCESS in 7m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c8d8bcc5e02044d5ac9c012f060fff7f : SUCCESS in 8m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e06975540e62443e88de388e51b40211 : SUCCESS in 8m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b6ab2546e2b94325a3ec6dd89ac0b669 : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5374dbfe17b9457db625353e4c815d43 : SUCCESS in 8m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/254d5f203c7d4d62910ef5c4daf5302d : SUCCESS in 1h 30m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49ab2fabbec949ac90c95e5934b4ccfd : SUCCESS in 1h 45m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8203ac47f740433193669b81cc2568df : SUCCESS in 1h 09m 23s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/713a0ec7743a4a3aaf816d8d40fefe37 : SUCCESS in 35m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1254f6ecaf7f4668a59089de53787f75 : SUCCESS in 33m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3125f976b51849ba987ec2e56bd5f091 : SUCCESS in 51m 43s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/af12fb9733b54a3bbd6861be830add5e : SUCCESS in 1h 32m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3d7eee4609044904b5b35b6219a3e352 : FAILURE in 2h 21m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dbc2252f37f9475ea84c9cdeb4ddddbc : SUCCESS in 1h 33m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a040fe88f6e64aa7ba65fb8baf4951d7 : SUCCESS in 1h 43m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/75e1327c576c4ab5877162dd946a9af7 : SUCCESS in 1h 36m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ab16d1a46dfa40fda37cc87374349b2e : FAILURE in 10m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66cbfa4c75fb9075d4bad35786938956e6913681","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-04 22:34:49.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"24d5c11df828745b85e6e1b984072bf7e7a83fcd","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-08 12:18:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"809c4508dd03460ad1c6fdc39cba109e098b6900","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 12:20:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"16ea3d646fe9dcff4ba80cf256183fd163a34214","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 15:02:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/216a22a47d544bbeb9211e16422d7729 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e22b020fde694d83814ef619acb26b23 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/803699c014a94bd781d2abeffd2a1d3f : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5112f105323447695f62c0f40058c18 : CANCELED in 10m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f005482dfa7e49d4800edd0749777ecd : POST_FAILURE in 8m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b0760d23edf6416cbd581e33ea535df8 : SUCCESS in 5m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2e90680e09044425adfa13a0b657102e : CANCELED in 8m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e2900f8ed4d9462b9cb2e49b399962d4 : CANCELED in 9m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee6bc387f13b4fcda12e38a23f137402 : CANCELED in 10m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/60a38c8409bb4700a53eb06ead28a32c : CANCELED in 10m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3a24a4b8a9084024b457a7c91a4501ce : CANCELED in 10m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ed4a033d0b143b1b7a6370169be41c5 : CANCELED in 10m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/537bd62ecf7f4af488ee9c4927d577ab : CANCELED in 10m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1c4c28a3605cb63fd55480978339f12bbccb62b4","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-09 08:10:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"667fee6649f76dae8cdfc349467f4a6062fb0d29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 10:14:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b368a3182de248d8a8c630fca6dcb2d1 : SUCCESS in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10e051d1d2fb453d90f068de0dc955ec : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b46110ebfafb42cab3c6d42903dfd691 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb9f88eb533f4b8f838a48a39fb3b805 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0753889cf0f84bd1b28950688655de4f : SUCCESS in 13m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/969b178058af459e8b7e9f781056afbb : SUCCESS in 6m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/738efc405aad4fd884c35127703270bd : SUCCESS in 7m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2f3ca4b14dae468cb0dba47144464747 : FAILURE in 13m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/eea05974db4f45d1b35fea605784b1d0 : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8f75813f08d74b35a26ef20804aa7e94 : SUCCESS in 8m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b0bda64728d347dd8b933c8cffc7de44 : SUCCESS in 1h 39m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3385633a5cb34decaa28c188e9b96cd3 : SUCCESS in 1h 24m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6ad91299ccef43528542176dba176f52 : SUCCESS in 52m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2fda008babf64dab9af6328519a70ed0 : SUCCESS in 32m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/70e71e546ca748af8a437776b684b06c : SUCCESS in 52m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b8e6c915be6e4e3faefec44458f2f574 : SUCCESS in 50m 39s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0675c959cc764b53a1d4aa58d83fdb55 : SUCCESS in 1h 30m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a1fbc53ceecc4304b29569a563c863dd : FAILURE in 1h 59m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b5fef5fbe1a94346bb9cca82bb9e0e8d : SUCCESS in 1h 58m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a642c1cf09ba48309ea8e4b86651faf4 : FAILURE in 1h 44m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a2b2ef2bc23741e98600e435110f3a5b : FAILURE in 1h 43m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4a67924462684e74a21301ef4c2e3ccc : FAILURE in 11m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aba61ec3de672c22e73921b9d87bf7f9548adcad","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-06-18 08:31:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"617f9fd1fac27701dc71a13b11de0d707cbd9534","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 10:50:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3442cbe0fce24a63bc1ac98b4b12e9cb : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b07959f2e0c4f2293cc6d909964ed27 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9af81fd17d474f0a962fd7be38ff3809 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e48e7f674e2475d97aa9a8bccc04631 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d66596cc1b4d4f1a9d34652d84ed25cf : SUCCESS in 12m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17b74193b29145c99dbf46fb421c48ff : SUCCESS in 6m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7141a16a25d546688b2be07c348d0fc5 : SUCCESS in 6m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ca86ade819d746d085fa17f2ef8acfea : SUCCESS in 9m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ed8672d52300444b813e3dfa0576bf13 : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a169d7ba7d684326b6f0ddeaa8cc9706 : SUCCESS in 8m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cc7672ae31334970a8997dcde4864367 : SUCCESS in 2h 04m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2678d605cd2c4b7f9077d49304c84de1 : SUCCESS in 1h 38m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/60034fd3bb8d4d2fa6194031dfa285d3 : SUCCESS in 43m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/845b01d0f0e54419b46d29f304d21591 : SUCCESS in 44m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a9784c5026ec41deb8d28cdd32dbc877 : SUCCESS in 40m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e7a76a6476a24f618c2fd4e642630bdf : SUCCESS in 58m 05s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/80abeeeb4ee94b1ebfaf984181368265 : SUCCESS in 1h 51m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a2964714d8934416adb0b1472758c0af : FAILURE in 2h 04m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d266f6e197dc4b5ab5b3261ff1efd5fc : SUCCESS in 1h 36m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4e3c6a8c8e8c448aaa34e0b5409bf6d1 : FAILURE in 1h 40m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5b840c432ccd409cbe624583d55650c5 : FAILURE in 1h 33m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0347464ade034bb2b48dc68d50ebc612 : FAILURE in 10m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"127633b3611116c15ec5dd2d12f846f415cb6700","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 10:51:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40f3fab69a96eece58409fb91b919c69ed236a9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 12:33:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1ef9f50ec71431f8689a98e7e6ca9bf : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94bd9d4ef0cb417cabc68bcdc110cc3e : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a71b6e0bdfae4d6e888808523ae0b694 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65baa75cd84340ec8bd8016a17eae0c0 : SUCCESS in 13m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fdebc1f179d4e9ebc2563651c1c59ca : SUCCESS in 6m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bce46cf7bd194bbe85f6459b34fdc815 : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/19ae4d6da1bb438b9daa1be6f3630617 : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2d0ce262fd294237b66c679cb602fdf4 : SUCCESS in 1h 36m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a9475b0a4594ea4ac9f23445d3fdbd3 : SUCCESS in 1h 33m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c1c9169deb4f4a86acd34d996ae76941 : SUCCESS in 43m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c2f8da40600a43bbbc64c5c7539d9b49 : SUCCESS in 43m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6d2a20557bf94e578d602c5f4daee9d3 : SUCCESS in 37m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ec7a36a8d51845b09743f0c660eec3f0 : SUCCESS in 59m 32s","accounts_in_message":[],"_revision_number":1},{"id":"0adc701371708ee3116cbb05f51591fe96239f1b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 12:33:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2b192294917223bf4f939da89a38d42560d7797c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 12:36:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcc3fe8562684cba859efac2e58c60e1 : SUCCESS in 2m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f30ffaa8d444f7a99f962e37872f7c4 : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2195a45a1ba14736211bd7305989e8ae9d5d9934","revisions":{"2195a45a1ba14736211bd7305989e8ae9d5d9934":{"kind":"REWORK","_number":1,"created":"2021-05-17 07:23:06.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/57/791657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/791657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/791657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/791657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/791657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/791657/1"}}},"commit":{"parents":[{"commit":"035d6d6daa86488a3a1da8a081a65ba9d21436c9","subject":"Merge \"Fix empty Batch Member Update to unlock objects\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/035d6d6daa86488a3a1da8a081a65ba9d21436c9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-27 09:14:43.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-05-17 07:22:53.000000000","tz":120},"subject":"Fix rsyslog configuration when disabling logs","message":"Fix rsyslog configuration when disabling logs\n\nFix an invalid rsyslog configuration when disabling both log offloading\nand local log storage.\n\nStory 2008782\nTask 42173\n\nChange-Id: I3125cfc7b3bd12eed139b058fc2bb7f23f12abc7\n(cherry picked from commit e26664de59eb872a85520c24b5701dae31cd75e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2195a45a1ba14736211bd7305989e8ae9d5d9934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2195a45a1ba14736211bd7305989e8ae9d5d9934"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
