)]}'
{"id":"openstack%2Foctavia~791660","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I3125cfc7b3bd12eed139b058fc2bb7f23f12abc7","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I3125cfc7b3bd12eed139b058fc2bb7f23f12abc7","subject":"Fix rsyslog configuration when disabling logs","status":"MERGED","created":"2021-05-17 07:30:56.000000000","updated":"2021-05-19 20:34:37.000000000","submitted":"2021-05-19 20:32:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791660","meta_rev_id":"a0b9635ace2c9ea18d24486365212da66c09e73d","_number":791660,"virtual_id_number":791660,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-19 20:32:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2021-05-19 17:46:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2021-05-19 17:54:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2021-05-19 17:54:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 07:30:56.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-05-17 09:57:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-19 17:46:40.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"fba663fec81ca640e0f932d8046288494974e3d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-17 07:30:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d26a9b6365f62b71233c49b4703d15a455975124","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 09:57:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2328abc2d52b4fe7a146af649ff020a4 : SUCCESS in 22m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5393f9578e344c9484f76941408d539a : SUCCESS in 19m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/700c94cdb601470d8dcc40bdc7f46dc2 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a90a26ffcc442b399d459d059774388 : SUCCESS in 17m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bdfcda2aca3f4433b8933eab19d900fb : SUCCESS in 17m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/250f54b403954421aedd756ebadd17b1 : SUCCESS in 11m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c77aa023562943538f734870f6dc230f : SUCCESS in 5m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/05276dbf09b44902bdacffc872f47fb8 : SUCCESS in 18m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a750e4ba5aeb4cf8927d81e3a34828e2 : SUCCESS in 19m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/52c3d958660440cea58abf750432a43b : SUCCESS in 8m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5decf92cfe1c4be08a60b4fa6cdbd50b : SUCCESS in 20m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/59ad3542e0e6486ea5c23576b8823285 : SUCCESS in 1h 37m 45s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c5174f4ebe0e4d409629f1332801985e : SUCCESS in 1h 34m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/63e7004366254a6d896616d4191e26cf : SUCCESS in 2h 05m 40s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b24ac125b9b54584a4e635d149627075 : SUCCESS in 1h 56m 03s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9fa541f477a342f38e398c40efc383a8 : FAILURE in 10m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/91b3a74a344a4927baf1b13277ce7e78 : FAILURE in 52m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/480c74f85bb6409cb3e0869df8d034d7 : FAILURE in 44m 50s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e42c9a10e1e34371a4340cabb951985d : SUCCESS in 56m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fd28679ecb784bda8a0addb526cd74db : SUCCESS in 1h 35m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8406d6fb39cc4226a4c5fb26c914cccb : SUCCESS in 39m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d51d2d8d16ce41d1a3740c794006ea0c : SUCCESS in 34m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a3106f0893ad43babe203a935ff2f5f1 : FAILURE in 10m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ce80e0b4a67650fe413ebe4cf7d5eb7285673c05","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-19 17:46:40.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"b8e4c29d89e17f935da167216bffe17776345e7d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-19 17:54:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2bf40fff86a6a308d8c63f734be4f0df7b9a7653","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 17:54:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3383914d7dfae0b95ee1b4d02f1fd0ed08e29ad7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 20:32:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5496ff2cb5b4e7a8ce133c2da6967a1 : SUCCESS in 18m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3dbc793a6cca47ad8f919081e804674e : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0be71c3147744e07944c4f447cc16b69 : SUCCESS in 16m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6612439d3611477cbad2b5ab183ece4a : SUCCESS in 16m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6108b9e1103146349abf61b0fe97bc16 : SUCCESS in 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ad54e16ef8249fab041963e8ba8ad43 : SUCCESS in 5m 24s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/05db0f2044624d72bc72fc896c6dd390 : SUCCESS in 7m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b26d10b0174b429f8e17d543d00897eb : SUCCESS in 19m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a6b5907241734787a09b6b520319e8ef : SUCCESS in 1h 45m 51s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/08ba043653674bf0b8f07b884f56520d : SUCCESS in 1h 44m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e3d51ddce06b4aed920ac7e9e7923c6c : SUCCESS in 2h 09m 18s","accounts_in_message":[],"_revision_number":1},{"id":"556f663675554e5ae3185871835f92f706823a02","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 20:32:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a0b9635ace2c9ea18d24486365212da66c09e73d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 20:34:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91564cb3b35f49eeb596e9b11ec0cb86 : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a89a202638a41f7952bb99cb4c05ad5 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de4514d6103cbe3babea57fae00e9ff5028f71b0","revisions":{"de4514d6103cbe3babea57fae00e9ff5028f71b0":{"kind":"REWORK","_number":1,"created":"2021-05-17 07:30:56.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/60/791660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/791660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/791660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/791660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/791660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/791660/1"}}},"commit":{"parents":[{"commit":"db24d7bb5187676c0501e722b7eaa5bd9161829e","subject":"Fix weighted round-robin about UDP listener","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db24d7bb5187676c0501e722b7eaa5bd9161829e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-27 09:14:43.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-05-17 07:30:23.000000000","tz":120},"subject":"Fix rsyslog configuration when disabling logs","message":"Fix rsyslog configuration when disabling logs\n\nFix an invalid rsyslog configuration when disabling both log offloading\nand local log storage.\n\nStory 2008782\nTask 42173\n\nChange-Id: I3125cfc7b3bd12eed139b058fc2bb7f23f12abc7\n(cherry picked from commit e26664de59eb872a85520c24b5701dae31cd75e9)\n(cherry picked from commit 2195a45a1ba14736211bd7305989e8ae9d5d9934)\n(cherry picked from commit 130bfcd4ae495168d18aea284cd38e40bf949029)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de4514d6103cbe3babea57fae00e9ff5028f71b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de4514d6103cbe3babea57fae00e9ff5028f71b0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
