)]}'
{"id":"openstack%2Foctavia~791667","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~Ie4d40509ae7e6faccecd1d0cc0bc4c7d93959261","project":"openstack/octavia","branch":"stable/wallaby","hashtags":[],"change_id":"Ie4d40509ae7e6faccecd1d0cc0bc4c7d93959261","subject":"Optimize CountPoolChildrenForQuota task in amphorav2","status":"MERGED","created":"2021-05-17 07:37:28.000000000","updated":"2021-06-08 14:41:26.000000000","submitted":"2021-06-08 14:38:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791667","meta_rev_id":"0904d511da8c9f98eb2dce45c35848953b23d698","_number":791667,"virtual_id_number":791667,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-08 14:37:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-06-08 12:17:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-06-04 22:50:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-06-08 12:17:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 07:37:28.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-05-17 10:30:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-04 22:50:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-06-08 12:17:40.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"c6862c8557115ad28280545f279dbae77194eec9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-17 07:37:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7bd4ae9406068d60cfd1e02e6b676a707f12c7eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 10:30:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6537edf8b95446baa3279f53c01c22f3 : SUCCESS in 10m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3e46e1ed24249afbaf4895bea21c1da : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58393a9590e5450f81853dbdbc6699c2 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7dd824cfbcf446ebf61e337420128c8 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/051fc08a255440df888fb723d8f0166e : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d08a311726554ef4b963614ba0d9e524 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b15d544d6a374f7aa224e1377fe34c8d : SUCCESS in 12m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f4032e133a52474a8debecfd74664cd8 : SUCCESS in 6m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6e14ccf12d2340a398de96f96b0698d0 : SUCCESS in 8m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0f8c0cedad7f40fb86276e49e65688de : SUCCESS in 5m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ac1e7fb2337b41c3a5dcd06bba5794f1 : SUCCESS in 8m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4c4bbdd5ebdb43cb96f287e449400e84 : SUCCESS in 1h 30m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1494de6a59e44d9ca89c9088668dfbe1 : SUCCESS in 1h 52m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ccb43271b93043c7b0e5a46fdef05531 : SUCCESS in 54m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ba3a076640e4412ba433a648388d0588 : SUCCESS in 55m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4691e594d3c24f9a8a10d6a18c20f141 : SUCCESS in 37m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5a16bdca63a7436f933c782aa7b58a6b : SUCCESS in 54m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1f26af2140fb4572835da60d87dd3769 : SUCCESS in 1h 46m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8e976649ca78497d82e17ddf7011f081 : FAILURE in 53m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/18cef090c9354e9a86806a8f6a1702b7 : SUCCESS in 2h 10m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/641e6bf36a684cff8f2c9d3ea5688373 : SUCCESS in 1h 56m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/31d939f0ded74aeaa68e336a42facccd : SUCCESS in 1h 33m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/97b2f281633a48cb95dca4849967c40d : FAILURE in 7m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a5387ab4846ca4f869372af4d9d9c560137da4a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-04 22:50:59.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"444368015953fddb6ddafa1e3dd8799dcafdd289","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-08 12:17:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"09fc4d7fb8768c5fb4809344c3eec9f35da03136","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 12:18:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4c89afd76366d980dcbab67ab2f24636ea7bebc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 14:37:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6edaa36d28f4490926b548fc17fb319 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/828108c49daa45c6bbb952f136329d1e : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0352ce8847c94cd3b73925d03bbbca3f : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf3d191ff5054dc8905c9349ad59c0c6 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e0e42be053d4505890a5f1fcae1f376 : SUCCESS in 12m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/eb10bd60d7064617ac560ce6c164a181 : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7aa152edf1cf43ddb8111b83b489be98 : SUCCESS in 9m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/65be3e8c31674b5e949f7bae00755a64 : SUCCESS in 1h 31m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/669388c5a74f40c08ef8b8895c0f6dc9 : SUCCESS in 1h 42m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5a889fa1479a4d55b1610caae555695e : SUCCESS in 43m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/979969572eea4fed927251350bf7477f : SUCCESS in 45m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e92dfa23f01546a38a9a9e05363abc6e : SUCCESS in 39m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b7eba143575b41358cf5fa02342ef602 : SUCCESS in 1h 06m 04s","accounts_in_message":[],"_revision_number":1},{"id":"50f6f6ae7870b39de52f7bda9cc43ce529aa92d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 14:38:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0904d511da8c9f98eb2dce45c35848953b23d698","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 14:41:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07060f8a280e42109ab67915b44fd1c6 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"adb454b2e8a9fc7b69c40c4fa62c21020a68524c","revisions":{"adb454b2e8a9fc7b69c40c4fa62c21020a68524c":{"kind":"REWORK","_number":1,"created":"2021-05-17 07:37:28.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/67/791667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/791667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/791667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/791667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/791667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/791667/1"}}},"commit":{"parents":[{"commit":"46979a0d1b92647913f693ee5b4d8e4b81473e7e","subject":"Cap hacking and pylint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46979a0d1b92647913f693ee5b4d8e4b81473e7e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-30 07:21:45.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-05-17 07:37:15.000000000","tz":120},"subject":"Optimize CountPoolChildrenForQuota task in amphorav2","message":"Optimize CountPoolChildrenForQuota task in amphorav2\n\nThe CountPoolChildrenForQuota task takes 3 or 4 seconds to execute in\namphorav2. When cascade deleting a load balancer with many pools (60),\nit might take minutes.\n\nWe don\u0027t need to get and to convert all the fields from the object to\ncompute the number of children of a pool, we can optimize this task by\ncreating a new method in the PoolRepository class.\n\nStory 2008873\nTask 42414\n\nChange-Id: Ie4d40509ae7e6faccecd1d0cc0bc4c7d93959261\n(cherry picked from commit 7075d22c4cb0dfc25b4c5c8eff654aa4358f2f25)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/adb454b2e8a9fc7b69c40c4fa62c21020a68524c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/adb454b2e8a9fc7b69c40c4fa62c21020a68524c"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
