)]}'
{"id":"openstack%2Foctavia~791668","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~Ie4d40509ae7e6faccecd1d0cc0bc4c7d93959261","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"Ie4d40509ae7e6faccecd1d0cc0bc4c7d93959261","subject":"Optimize CountPoolChildrenForQuota task in amphorav2","status":"MERGED","created":"2021-05-17 07:38:01.000000000","updated":"2021-06-24 19:25:59.000000000","submitted":"2021-06-24 19:24:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791668","meta_rev_id":"69d9b52059b810436dee0bf8be73f8a5502126ce","_number":791668,"virtual_id_number":791668,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-24 19:24:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-06-24 12:25:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-06-22 09:51:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-06-24 12:25:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 07:38:01.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-05-17 10:54:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-24 12:25:38.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"902a807e685cc3394b4368b81e23f82dfc6671e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-17 07:38:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac248af779540dbed32fb58713069d84cd0f36f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 10:54:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fab8a98c987545169ba7484be6131f99 : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7408500729454ea9a7f4364c84710bea : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d1998a0dba847ea9f8a7c3876a993f1 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/860cc865bde449dba476260c24d4a5d8 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beeb2e88bdad47bfa787a64f26606a39 : SUCCESS in 14m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3b15e553ff1a4c43b76e6214e235fc1e : SUCCESS in 6m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/68ae8ef015204bc090b54d440b1bce1d : SUCCESS in 9m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ee23bbed535e406599ab7ea00887beaf : SUCCESS in 5m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b25c16ddf14e4a64b9981ffcb8a4e7a3 : SUCCESS in 10m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3e9022c5a55c409d8b1ccea2d30dda91 : SUCCESS in 1h 45m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b1216af592914763b4b4ddb3b17a3c04 : SUCCESS in 1h 54m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b7614191e2574b868778c7222c5dfff8 : SUCCESS in 44m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/aab396a8b2384d9cb846f7b093b45804 : SUCCESS in 34m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95fcb7689ed84923868345520da6fbd9 : SUCCESS in 35m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7fcff3647b954efb93cc6207dcf5097d : SUCCESS in 50m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/addea3374b9748a78d47ce48cb7af51b : SUCCESS in 1h 54m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/80d98bda71cc42c4a87265046fdf1763 : FAILURE in 2h 04m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/334728da3ee2426caf6df45146712b72 : SUCCESS in 2h 25m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b90b0327f96f455f9f483c128a8c9668 : SUCCESS in 2h 06m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e92f4fcfad424564924fdff5b22887d4 : SUCCESS in 1h 51m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/aac12602745f4f539025ba5790f76583 : FAILURE in 8m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"edebbab6fdb0c5be235a67b2a74a924ad2a500ba","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-22 09:51:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"80e7575a756d8f6511e9ea703a2592f34c9aaf14","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-24 12:25:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"27659c7d231c17e1d4e5b804ae26a5d80ae0e3bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 12:25:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5e5c9db481db31f75ad68bfc39416095fa075b8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 17:20:13.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dde5ceac3acec560ba8b72c87d95e6567e2b4c10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 19:24:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd7a29c755c74f228f896b5d93a75231 : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4816e388bde148bcb82384d4eeb6a453 : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a198602077a43d3aef6902b9d5b6484 : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca02d819e915498db639fe09eacad3d5 : SUCCESS in 15m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1892f720e6d64533bc95e7c1c80cf2be : SUCCESS in 8m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e1716f2e020442d592efe1778b97defd : SUCCESS in 11m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/889963a744a244688ca0b76c665eb58b : SUCCESS in 1h 41m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0903c6dd8e0e4579a490c278dc4fdd90 : SUCCESS in 2h 02m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/25d3b9e330564a42af193b9fe3b47668 : SUCCESS in 43m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8a7685606d77457994b90630257fcd2b : SUCCESS in 1h 05m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95a999b1aef9458cbe474cab46bd8eee : SUCCESS in 35m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ae9ba5dac10b43e29d61db23dc824056 : SUCCESS in 1h 07m 00s","accounts_in_message":[],"_revision_number":1},{"id":"516bfa97710db5cedc911aa3be619687391baeb5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 19:24:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"69d9b52059b810436dee0bf8be73f8a5502126ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 19:25:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/318a6296becf45389e773cb934d40d38 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37b15da4a13da800b991d81e89e15fe1ff216568","revisions":{"37b15da4a13da800b991d81e89e15fe1ff216568":{"kind":"REWORK","_number":1,"created":"2021-05-17 07:38:01.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/68/791668/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/791668/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/791668/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/791668/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/791668/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/791668/1"}}},"commit":{"parents":[{"commit":"035d6d6daa86488a3a1da8a081a65ba9d21436c9","subject":"Merge \"Fix empty Batch Member Update to unlock objects\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/035d6d6daa86488a3a1da8a081a65ba9d21436c9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-30 07:21:45.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-05-17 07:37:49.000000000","tz":120},"subject":"Optimize CountPoolChildrenForQuota task in amphorav2","message":"Optimize CountPoolChildrenForQuota task in amphorav2\n\nThe CountPoolChildrenForQuota task takes 3 or 4 seconds to execute in\namphorav2. When cascade deleting a load balancer with many pools (60),\nit might take minutes.\n\nWe don\u0027t need to get and to convert all the fields from the object to\ncompute the number of children of a pool, we can optimize this task by\ncreating a new method in the PoolRepository class.\n\nStory 2008873\nTask 42414\n\nChange-Id: Ie4d40509ae7e6faccecd1d0cc0bc4c7d93959261\n(cherry picked from commit 7075d22c4cb0dfc25b4c5c8eff654aa4358f2f25)\n(cherry picked from commit adb454b2e8a9fc7b69c40c4fa62c21020a68524c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37b15da4a13da800b991d81e89e15fe1ff216568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37b15da4a13da800b991d81e89e15fe1ff216568"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
