)]}'
{"id":"openstack%2Foctavia~791669","triplet_id":"openstack%2Foctavia~stable%2Fussuri~Ie4d40509ae7e6faccecd1d0cc0bc4c7d93959261","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"Ie4d40509ae7e6faccecd1d0cc0bc4c7d93959261","subject":"Optimize CountPoolChildrenForQuota task in amphorav2","status":"MERGED","created":"2021-05-17 07:39:27.000000000","updated":"2021-06-17 22:40:28.000000000","submitted":"2021-06-17 22:40:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791669","meta_rev_id":"b7b4d07793fe520a2760f5d8164e7824b5d47d97","_number":791669,"virtual_id_number":791669,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 22:40:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-06-17 22:40:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-06-17 22:40:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-06-17 20:37:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 07:39:27.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-05-17 10:35:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-04 22:50:00.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-06-08 12:17:57.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"993be30cef5c5d97c6caccafe5bea2f3be302399","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-17 07:39:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ebe0158c154b0d14bc0d985d2732398197d0e91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 10:35:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0360fbf81ea04246933c2027b83bd611 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba3629e9c5e845cf98fcf8915cd00aa5 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fbaf5611a84479faaf85ce0ea20bebf : SUCCESS in 6m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09b4b35f2f1548c39fd3770ab493fde5 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c02f4bc1b1b9402b8098dd57eac0acca : SUCCESS in 12m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0c7785af0662421380a1cea6eae283d9 : SUCCESS in 6m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/024d90c2578f46c89a305c029859a293 : SUCCESS in 8m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aa576a3b288d4b7fbddd4559f34692a1 : SUCCESS in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6f24b300998d4f57a977eccdc5aeb5ca : SUCCESS in 1h 33m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/327b1a79f8ce46f68d147f44fd8d1163 : SUCCESS in 2h 00m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/48eb4dba2bb547f095852c7bbde52bbd : SUCCESS in 42m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7ad12cd0c1db48e8a80995c84ee5f32f : SUCCESS in 46m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab1a7a8b29694e0ebac62e48bb9e1830 : SUCCESS in 43m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e2866389d1814b5a857f7596541aa421 : SUCCESS in 1h 12m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/41677fee2e994bf580accf51733a8144 : SUCCESS in 2h 00m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e47abb805f1341bbaead8fa7ec4b9577 : FAILURE in 47m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6d7ea4df91134039a79adb70973490f8 : SUCCESS in 2h 06m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2cff265c327245c7a4032c7b516d8a89cea61b1f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-04 22:50:00.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"41b588526f430aaa396b3706f45ba03e9cf9efa9","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-08 12:17:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1eb2e359c7b790477dfabcfa3a391436a5b1fac0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 12:18:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ddf6bac8b6cbeb5548a5000c6cb7d17501c5d46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 14:38:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/276246d5e72341098f005ad95b64fb4a : CANCELED in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9200f2de90ea4c4589405a4079e1c8be : CANCELED in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/add6c51ddaec424aac83e6c925275366 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e7f99e849f64d07833a549bd23f695d : CANCELED in 6m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e36960e559de468ca297e325f5b9cfce : CANCELED in 6m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ad47efe02aa443cea01194e0efb59542 : CANCELED in 6m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79386ebfd4e943bf9e843b24f2aec94c : CANCELED in 8m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d4fc0087a9ea47a0b86df55e39a7ea2a : CANCELED in 8m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3afb33c44d3a46138049e34bc94eb334 : CANCELED in 8m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca2638e8334c46a69084a1272a750346 : FAILURE in 8m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/14341192255e483abf1f30c83196f987 : CANCELED in 6m 04s","accounts_in_message":[],"_revision_number":1},{"id":"fa1e41c96b3a5bf236731bab19b5951fcb5ccaab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-06-08 14:48:41.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"8df15d20afacd151d512c1b3e2e618b8524f54b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 17:40:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f1509b068dc4708a112ff0ccdda4dd4 : SUCCESS in 12m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f65c56f294584b42ae835084a8155dcd : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f457db0f6a244115a00bc27f670463fb : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b2f41679342b4fee8d4f454edfd15032 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d11469dd279447d98a12b3a5bdd46a3 : SUCCESS in 14m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/846f57f4bfb44089acc8f6eef9829c43 : SUCCESS in 7m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3e6d0b3ce0ec4ca88b0e910b0293b4b0 : SUCCESS in 8m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/93db6cbe6cb44df09b086c2750ef010f : SUCCESS in 7m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/00e77cc3e9eb4a96888de6de92a45966 : SUCCESS in 1h 38m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/867813cb8dcf44db9982c3a17a1c7a3d : SUCCESS in 2h 01m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e1612cce7a25426783605a1f05e7750f : SUCCESS in 43m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e45c9c02835b4c518efec8210577b4f5 : SUCCESS in 40m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/388ce385b47a49b8a05b2377c9361749 : SUCCESS in 43m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3c4eef0f17ff499487173cbc24c6a5f4 : SUCCESS in 58m 35s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4128bf8c483b49b09a929e67a9c0cfe8 : SUCCESS in 1h 59m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/57fa9317c982408388d949b6cafcc3d9 : RETRY_LIMIT in 7m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/488ab2e5863842c78e5534ca06c7ed5a : SUCCESS in 2h 09m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5df2b6e11a65d01700ea7a3d755915f7faf17960","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-17 20:37:29.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8c851ab9624d3c417df4e85f28aa02ca2a1a341f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 20:37:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a95da5aa03f717399a9fad3bc87f63051757be14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 22:40:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b495d45f6714b968e3eda57dabd191b : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecfd63241e8d4ae081ac3e733cb7dd4d : SUCCESS in 5m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2541e4e14957458b897df73a3a7e0a7b : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f7cf57626dc4cdf9ea8ef4238cc539f : SUCCESS in 12m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c21a7cdb85c94a7380daa01cd8268e14 : SUCCESS in 8m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fc5828af8f0449de99d710209c3eb8d7 : SUCCESS in 1h 22m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a438d536efec46358735022ada47cb6a : SUCCESS in 1h 54m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0e9655a314404fc2b058aeabfef4572d : SUCCESS in 35m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/27093285111944cfa45c2cf60b0b32c8 : SUCCESS in 39m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/80aefadbd87e4fd9a012076aa734c364 : SUCCESS in 39m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9125855f93894534a0f1fe66228ae93f : SUCCESS in 1h 03m 45s","accounts_in_message":[],"_revision_number":2},{"id":"1f94d33505fe489ff0e9c0854cfb6b75d21c0d1b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 22:40:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4cfbf6669d77c78b95fa7252b5032e33db4bd1a1","revisions":{"5000ac4c9a354e6cf4b16c98bba04da361c5e45f":{"kind":"REWORK","_number":1,"created":"2021-05-17 07:39:27.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/69/791669/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/791669/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/791669/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/791669/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/791669/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/791669/1"}}},"commit":{"parents":[{"commit":"46b9ce7cde29cce0a8cf94fcb2ca8587d2a71bdf","subject":"Cap hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46b9ce7cde29cce0a8cf94fcb2ca8587d2a71bdf"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-30 07:21:45.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-05-17 07:39:13.000000000","tz":120},"subject":"Optimize CountPoolChildrenForQuota task in amphorav2","message":"Optimize CountPoolChildrenForQuota task in amphorav2\n\nThe CountPoolChildrenForQuota task takes 3 or 4 seconds to execute in\namphorav2. When cascade deleting a load balancer with many pools (60),\nit might take minutes.\n\nWe don\u0027t need to get and to convert all the fields from the object to\ncompute the number of children of a pool, we can optimize this task by\ncreating a new method in the PoolRepository class.\n\nStory 2008873\nTask 42414\n\nChange-Id: Ie4d40509ae7e6faccecd1d0cc0bc4c7d93959261\n(cherry picked from commit 7075d22c4cb0dfc25b4c5c8eff654aa4358f2f25)\n(cherry picked from commit adb454b2e8a9fc7b69c40c4fa62c21020a68524c)\n(cherry picked from commit 37b15da4a13da800b991d81e89e15fe1ff216568)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5000ac4c9a354e6cf4b16c98bba04da361c5e45f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5000ac4c9a354e6cf4b16c98bba04da361c5e45f"}]},"branch":"refs/heads/stable/ussuri"},"4cfbf6669d77c78b95fa7252b5032e33db4bd1a1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-06-08 14:48:41.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/69/791669/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/791669/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/791669/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/791669/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/791669/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/791669/2"}}},"commit":{"parents":[{"commit":"d5dc2b260169611ccc0f9d8193369ca15515b8cb","subject":"Fix task_flow.max_workers with persistence in amphorav2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d5dc2b260169611ccc0f9d8193369ca15515b8cb"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-30 07:21:45.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-06-08 14:47:47.000000000","tz":120},"subject":"Optimize CountPoolChildrenForQuota task in amphorav2","message":"Optimize CountPoolChildrenForQuota task in amphorav2\n\nThe CountPoolChildrenForQuota task takes 3 or 4 seconds to execute in\namphorav2. When cascade deleting a load balancer with many pools (60),\nit might take minutes.\n\nWe don\u0027t need to get and to convert all the fields from the object to\ncompute the number of children of a pool, we can optimize this task by\ncreating a new method in the PoolRepository class.\n\nStory 2008873\nTask 42414\n\nChange-Id: Ie4d40509ae7e6faccecd1d0cc0bc4c7d93959261\n(cherry picked from commit 7075d22c4cb0dfc25b4c5c8eff654aa4358f2f25)\n(cherry picked from commit adb454b2e8a9fc7b69c40c4fa62c21020a68524c)\n(cherry picked from commit 37b15da4a13da800b991d81e89e15fe1ff216568)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4cfbf6669d77c78b95fa7252b5032e33db4bd1a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4cfbf6669d77c78b95fa7252b5032e33db4bd1a1"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
