)]}'
{"id":"openstack%2Foctavia~791670","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I99ce408ec886820c056b69696b26be9521740f1c","project":"openstack/octavia","branch":"stable/wallaby","hashtags":[],"change_id":"I99ce408ec886820c056b69696b26be9521740f1c","subject":"Fix comment for the ca_certificates_file opt","status":"MERGED","created":"2021-05-17 07:40:56.000000000","updated":"2021-06-08 16:53:35.000000000","submitted":"2021-06-08 16:51:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791670","meta_rev_id":"50a51c60496f713f9a6a94a2b04f2d47f76eb9ce","_number":791670,"virtual_id_number":791670,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-08 16:51:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-06-08 12:18:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-06-04 22:33:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-06-08 12:18:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"reviewers":{"REVIEWER":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 07:40:56.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"REVIEWER"},{"updated":"2021-05-17 10:53:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-04 22:33:32.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-06-08 12:18:51.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"7ed9fb9ba3c9fbba0a0f9342a8bc4cd01f74d51c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-17 07:40:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17b82075f6834cd457e15c5034bfb50a60a7934b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 10:53:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49a5d1ebdec8488ea68ad10e66f91191 : SUCCESS in 10m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47a39e616a1940f0b2398239b199e478 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3e31e975e4c42ec86f750e6d9c6b431 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee38d3ff71c44aefbf4cda505303881d : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4450e73acbf42f784fc901130fcc4f0 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa0feb2a5bc24c0d8859075b7da8a20b : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fd72c9223c0424e9fb75afcbb3d1416 : SUCCESS in 13m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ddb1795e63694d2481e3a2e59feb3b18 : SUCCESS in 7m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fb7e6a1bc7d84b5f93fe8327192f7bac : SUCCESS in 8m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e0e51c98e03a423baa16c6839e3331b3 : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f0a9b3e7967a4e0186b2609a10156ba7 : SUCCESS in 8m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5eb245cefa3c4a62bfc8557d312d9077 : SUCCESS in 1h 31m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/efc95d4883d44ef2854041328130f8ea : FAILURE in 1h 21m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eedabc2998bc4432946cae8a68015c9b : SUCCESS in 46m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5e285eaed46a43f1a445e1bdacad5b65 : SUCCESS in 31m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d8560087ee44fdbaa256e25490ab857 : SUCCESS in 34m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ff8e99272b0f435781ce495d232029e5 : SUCCESS in 50m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5f8f0b9152fc4368b5cb2d1f8196adcd : SUCCESS in 1h 38m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ec46331779624013a60429d7227832b8 : FAILURE in 45m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a16aea1bd9cb4e74a6698da9c0cade2e : SUCCESS in 1h 44m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f107558ae551427ab08712ac2b643b73 : SUCCESS in 1h 54m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/34002d11f4774f738c6393209a00641b : SUCCESS in 2h 14m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ee171176f51d4a789a591307b9da4140 : FAILURE in 9m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b804a06f6dc2b54729c41dbba2c8cb521ccaada3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-17 10:58:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bd174a1422e24a5bd84fe214ffc96ae900ce68ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 12:53:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb3391e2155d4e6bb4fc77ff705c26c8 : SUCCESS in 10m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f91a341f2e8d4c62a12e61599ddced76 : SUCCESS in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8420677185a84d669ae1d8d5de376f06 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b97a2813fc3d4e3e9fd0bdd7dc86210d : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec9a702fbfee485f9fbbbc9783042d02 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/274cc46d341a4ec48de699f5d4f513e2 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26b6e604cc484ce0b255c907aa29643b : SUCCESS in 14m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/05d3a13f45de475bb05f99f113d7d50b : SUCCESS in 8m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3e6b02a1f97f4f73858b9a9be7229049 : SUCCESS in 8m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1ce2f2d7596344caa1109cca7870393d : SUCCESS in 4m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/56abe209cf204e08b957d75d242c74be : SUCCESS in 7m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c4a18c56af68499d8ae0bb9883874418 : SUCCESS in 1h 12m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f01d37eb2bee4a19b9426cf073051a48 : SUCCESS in 1h 44m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8c1d2b44d9114d2f83e73ca7a336b3da : SUCCESS in 59m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9157e24640af4863979f537354bb49b8 : SUCCESS in 56m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f46f1c7d550b4721b76dfb0580aa4830 : SUCCESS in 56m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2c889eab3d714d90b0ceac2805a78477 : SUCCESS in 54m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/40b31dd0af5d4920b0ab2f30f380e12e : SUCCESS in 1h 23m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/61bac72525354baba678fda501032743 : FAILURE in 56m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4f3d534c8ca743f483280bd1612d93a8 : SUCCESS in 1h 39m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f6958c5c3cad4d92977e4c707112804b : SUCCESS in 1h 26m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c2785eb7b9eb41fb8f957cd3b9fcf68a : SUCCESS in 1h 26m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/66ede9aeaa7e438081d096f4389bd347 : FAILURE in 7m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4ba78e01611e6cc37175488e4aa0989dc383e57","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-04 22:33:32.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"96ee9798164a0b0e3d85895ce09e0ac862ce8072","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-08 12:18:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b1b4935a17d035efa6f6210a41c7534412d3ed56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 12:20:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52ada4e780bb4d41de4f450c0fc114210de02816","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 16:51:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95a1774ccbdc406eb7acd90f6b52c92c : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c533a76831542c59cd14c8b2b2b310f : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/183432f42df44d4da83089d5eede6a8c : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bba27d53aad4438aa1130a76c593253d : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/770f3225b1b846cca09b72493eb9ca3d : SUCCESS in 12m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/50c73a77ad954156aa7c4b9b44a1958b : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d2914e40a7f64567abb8264b3c421d0f : SUCCESS in 9m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0f27ced803644b54ac10a386e8cb4bda : SUCCESS in 1h 41m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/615a2b85a77144a89a2d6b579f315bfe : SUCCESS in 1h 35m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39c5213983ca45acb1411057eb34cbca : SUCCESS in 40m 27s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f3f4a0e2260a4f6e869f8b16c7b0beaf : SUCCESS in 42m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2f7b97631117473faaf4cec664aac898 : SUCCESS in 41m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/226b0d48b2814000ae5cfcf7c594436c : SUCCESS in 1h 02m 41s","accounts_in_message":[],"_revision_number":1},{"id":"23ddc2a5eaacd14b6768b192b1e10e7b47126984","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 16:51:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"50a51c60496f713f9a6a94a2b04f2d47f76eb9ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 16:53:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7521843cdf3a46b3946d37abe36de232 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20db85fd5236a4daed7ab29931fed91da2295ab0","revisions":{"20db85fd5236a4daed7ab29931fed91da2295ab0":{"kind":"REWORK","_number":1,"created":"2021-05-17 07:40:56.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/70/791670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/791670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/791670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/791670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/791670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/791670/1"}}},"commit":{"parents":[{"commit":"46979a0d1b92647913f693ee5b4d8e4b81473e7e","subject":"Cap hacking and pylint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46979a0d1b92647913f693ee5b4d8e4b81473e7e"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-01 12:11:07.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-05-17 07:40:41.000000000","tz":120},"subject":"Fix comment for the ca_certificates_file opt","message":"Fix comment for the ca_certificates_file opt\n\nThe directive [certificates]/ca_certificates_file currently has a\nconfusing comment. This tries to fix it and make it more easy for\nOctavia operators to configure the directive.\n\nChange-Id: I99ce408ec886820c056b69696b26be9521740f1c\n(cherry picked from commit ee0da827b163c15796bf2f19b0785bc0259d5128)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/20db85fd5236a4daed7ab29931fed91da2295ab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/20db85fd5236a4daed7ab29931fed91da2295ab0"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
