)]}'
{"id":"openstack%2Foctavia~791672","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I99ce408ec886820c056b69696b26be9521740f1c","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I99ce408ec886820c056b69696b26be9521740f1c","subject":"Fix comment for the ca_certificates_file opt","status":"MERGED","created":"2021-05-17 07:41:59.000000000","updated":"2021-06-17 22:40:37.000000000","submitted":"2021-06-17 22:40:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791672","meta_rev_id":"7872bbd6865e0210627d28391e720dc45e8c247b","_number":791672,"virtual_id_number":791672,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 22:40:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-06-17 22:40:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-06-17 22:40:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-06-09 07:49:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"reviewers":{"REVIEWER":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 07:41:59.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"REVIEWER"},{"updated":"2021-05-17 10:37:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-04 22:42:43.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-06-08 12:18:04.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"736d1631b5a77f53a6e68048e3291fd86fc3bcb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-17 07:41:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc19083af3cd3086b70c47c07cfe24c78825780c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 10:37:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05a45fc9b2254a89a85f443a79633d0b : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c58e4a5b04914cbcbe1131f5bedc8bd6 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cc7ab26ca654dfd89c03214ff6364fe : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f2d16ddaf9a46bca9471c6b02d3b538 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d497e582d2e94b2d89960e5f09fcf974 : SUCCESS in 13m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f80f77aaf9904d48b0f89eed2e62a480 : SUCCESS in 6m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1f25ebe466c54330a7d80cf9d7f3a3b1 : SUCCESS in 8m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d9390f16ce434a11be808c9e32a462d2 : SUCCESS in 7m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/26f930f9d12a47e49886f0be41817efe : SUCCESS in 1h 33m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8d851967dc684e64a4ddf9e614841650 : SUCCESS in 1h 42m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b4f401caf95d4e6087bf1170e903ea77 : SUCCESS in 42m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bd47324c289e4565a21d4b9229b31925 : SUCCESS in 36m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e7d6e265df3457aa522bb48dccbcb5b : SUCCESS in 53m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bc2c312c55464ef8a4486d9de51bbcb5 : SUCCESS in 1h 22m 02s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/19872cb5c0dd40a4aa98046c80697d63 : SUCCESS in 1h 53m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/017e1cd2e1b74955be6f77c4c24517ea : FAILURE in 42m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2cad1023185945dc9ce2b142a715d7bd : SUCCESS in 1h 33m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4cfdacb36d9ea920bb0097c1d82e386763a56afe","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-04 22:42:43.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"c8bb486c9024d256f7ef7de196ee8590ee0a3537","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-08 12:18:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"42a3addc0b590ce379ac9f9d8fd08f1557273d6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 12:19:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c9c2a5c405c272edf4ad814f10e59bf35c2f062","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 14:38:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8937c2463f747df9002fc258be07067 : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73c0ae88af6544bab64b9cb29c4d446f : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ea3644631ac4fc9be7ea6574687b2bd : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46e33e5a96404441b84e2bd57ebc148e : CANCELED in 7m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d793e2a06d6341e0a692097cbf4c044b : CANCELED in 7m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4dcf83e6368741d28f4ec28baf8734e8 : CANCELED in 9m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ebf76e1a6ac642128d032f186330d17b : CANCELED in 9m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9efef49c9de947cb826e2b0acdb95f06 : CANCELED in 9m 23s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4512f8b482e94052927fde13561fe4fb : CANCELED in 9m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/576bc88e625d4bf78fe12430576162ba : FAILURE in 9m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/909c61db8f9249ccb8f801985c7c7db7 : CANCELED in 9m 10s","accounts_in_message":[],"_revision_number":1},{"id":"2a0a96a433dac44601ca94b5745655d96a2c84e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-06-08 14:48:41.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"34a51417de0ca0d3be8a235d277c96f0791d1203","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 18:08:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/989b5d90823f4064be2efcf48dee332f : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6237345c6735483b81abdd9237610e81 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90a81eb835724f8fae1c39b3e127c6ca : SUCCESS in 6m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2590aa2f0e694bf7a26f85ed363de413 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bf9390f320e40629febf012a971ee08 : SUCCESS in 12m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ab3633ff946845dcbffbc93601b721ee : SUCCESS in 7m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c24fee18fdf042f2ad65b6d805788d8d : SUCCESS in 9m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/652b6098790c4526a3e82a58ae433d89 : SUCCESS in 8m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2bb8fcb5729544c6bcde3ef1d185fde5 : SUCCESS in 1h 38m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b74eee44be62427cbfbc41067b53e0ee : FAILURE in 1h 36m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8882351c98f54445936f046b5b6223d7 : SUCCESS in 44m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/fad8839bdce34b3f858a03a1b5f34c5a : SUCCESS in 1h 27m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3fa361d5f3b04003a2141288feb78b3e : SUCCESS in 40m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bb7c47505da04e1ebe7f7f132be2972b : SUCCESS in 1h 11m 07s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9df587895ba5487ead297f5d86a7fd04 : TIMED_OUT in 2h 35m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a5814f9474f2482db70996f0b409daef : RETRY_LIMIT in 7m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/95ef0ffd781e4b54a43e01ffef7b216e : SUCCESS in 1h 33m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d95c81e34e91a2417458acb5bf3fc3c814bb8fc0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-06-08 18:59:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b5d7825e750048be4e00c0f3be91c71603f5ba54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 21:03:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4be6b494081243c482c07afefe2c9f24 : SUCCESS in 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a97f7103c1b74f3a8aa0ece3d7628241 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/236aac1782b4444fa88bf7203f375ebe : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17ca762ba7b6433297ae543d31fe474b : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0beb585d60514fd39fa02ab1e1f05f9c : SUCCESS in 12m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d8bafcc62dc445529a38788a795f84d2 : SUCCESS in 6m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/99f54d2953e34cdfba48e3c0dd415f27 : SUCCESS in 8m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/89f870247354457ea0a62c300e048d20 : SUCCESS in 7m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4a4b14ec07a3411199a8e5d0aa29ac05 : SUCCESS in 1h 26m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/64fc12072e8a472a890370394c12b4b2 : SUCCESS in 1h 36m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fa3ec9bcb41d4d39893b8c915c0c5359 : SUCCESS in 1h 14m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/daeda5c047b840c8959a85d7289e9535 : SUCCESS in 38m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a6224969dadd4a90a8fcb936d4112aef : SUCCESS in 41m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/97e790b4378e4171bede9f3611fd3d0f : SUCCESS in 1h 10m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d1c8304fb30f4c9ba5dd8553cf72d6be : SUCCESS in 2h 02m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/34d87df863b44ba5b4d028006cab64eb : RETRY_LIMIT in 20m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/70a51e64d07f483db67794490597ce6f : SUCCESS in 1h 55m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7ff599d11a21a40ef1b4d6be4543dc49a75a027d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-09 07:49:24.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2d3d7d10580a5d31275eb03cff3c75f34970b22c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 20:37:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6ffa39b8599deaa4533556b940c82fa66bd7a754","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 22:40:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42612ca5744540949998af3b46a1fb7e : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2254f52875047498e9f12705989026e : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/305ed84dedce435a9f48bd00734fd843 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bda5be3586094801a59b72f9958e55af : SUCCESS in 13m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/55e3f0a77a9c4aee85ef8a07865f2bd2 : SUCCESS in 8m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/62b046cfc49c4a7098de8624881a8c7b : SUCCESS in 1h 18m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/26dddd1440bb40d59271837643a2faba : SUCCESS in 1h 43m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/022dfb3834294fb2a976ae8d40966783 : SUCCESS in 39m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c287ba8741f748b8b7e4a82f46417646 : SUCCESS in 39m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3deb0ab4e9044baeb5459a04c90a05d0 : SUCCESS in 33m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/78631076cc57428181811637f923fe4b : SUCCESS in 1h 02m 44s","accounts_in_message":[],"_revision_number":2},{"id":"cd2c58c35ee7e696ad946517f6c249e58b2b3663","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 22:40:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"45b31c18240b6d138a803b4e22163b6654554448","revisions":{"a8977b01ad85d06039956744895262629584f425":{"kind":"REWORK","_number":1,"created":"2021-05-17 07:41:59.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/72/791672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/791672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/791672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/791672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/791672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/791672/1"}}},"commit":{"parents":[{"commit":"46b9ce7cde29cce0a8cf94fcb2ca8587d2a71bdf","subject":"Cap hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46b9ce7cde29cce0a8cf94fcb2ca8587d2a71bdf"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-01 12:11:07.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-05-17 07:41:49.000000000","tz":120},"subject":"Fix comment for the ca_certificates_file opt","message":"Fix comment for the ca_certificates_file opt\n\nThe directive [certificates]/ca_certificates_file currently has a\nconfusing comment. This tries to fix it and make it more easy for\nOctavia operators to configure the directive.\n\nChange-Id: I99ce408ec886820c056b69696b26be9521740f1c\n(cherry picked from commit ee0da827b163c15796bf2f19b0785bc0259d5128)\n(cherry picked from commit 20db85fd5236a4daed7ab29931fed91da2295ab0)\n(cherry picked from commit d69d06660633da7370604a45f43e0014d023f9d9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8977b01ad85d06039956744895262629584f425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8977b01ad85d06039956744895262629584f425"}]},"branch":"refs/heads/stable/ussuri"},"45b31c18240b6d138a803b4e22163b6654554448":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-06-08 14:48:41.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/72/791672/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/791672/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/791672/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/791672/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/791672/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/791672/2"}}},"commit":{"parents":[{"commit":"4cfbf6669d77c78b95fa7252b5032e33db4bd1a1","subject":"Optimize CountPoolChildrenForQuota task in amphorav2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4cfbf6669d77c78b95fa7252b5032e33db4bd1a1"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-01 12:11:07.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-06-08 14:47:55.000000000","tz":120},"subject":"Fix comment for the ca_certificates_file opt","message":"Fix comment for the ca_certificates_file opt\n\nThe directive [certificates]/ca_certificates_file currently has a\nconfusing comment. This tries to fix it and make it more easy for\nOctavia operators to configure the directive.\n\nChange-Id: I99ce408ec886820c056b69696b26be9521740f1c\n(cherry picked from commit ee0da827b163c15796bf2f19b0785bc0259d5128)\n(cherry picked from commit 20db85fd5236a4daed7ab29931fed91da2295ab0)\n(cherry picked from commit d69d06660633da7370604a45f43e0014d023f9d9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45b31c18240b6d138a803b4e22163b6654554448"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45b31c18240b6d138a803b4e22163b6654554448"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
