)]}'
{"id":"openstack%2Foctavia~791673","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I99ce408ec886820c056b69696b26be9521740f1c","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I99ce408ec886820c056b69696b26be9521740f1c","subject":"Fix comment for the ca_certificates_file opt","status":"MERGED","created":"2021-05-17 07:42:27.000000000","updated":"2021-05-19 20:29:09.000000000","submitted":"2021-05-19 20:27:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791673","meta_rev_id":"7d4be689f3924eab4195761189deac06b8bebad3","_number":791673,"virtual_id_number":791673,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-19 20:27:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2021-05-19 17:47:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2021-05-19 17:53:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2021-05-19 17:53:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"reviewers":{"REVIEWER":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 07:42:27.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"REVIEWER"},{"updated":"2021-05-17 10:49:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-19 17:47:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-05-19 17:53:49.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"5eae19844db80ce359ede738ed7a87f94322407f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-05-17 07:42:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cee57ec0b71fd8de85d37a974d18c1793bd1533b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 10:49:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/156eaea09af842c68fb51590a8cff1ea : SUCCESS in 19m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbb0e43159ee4e96b1963304d2413518 : SUCCESS in 17m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d14a782efded472abf73458c57a49fe5 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7239ab32f5824f46ba4a821881ae754a : SUCCESS in 17m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2dfba37995674d469ae26e56d278fb35 : SUCCESS in 16m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2b51f7f32fa45a7bc9352c207cff685 : SUCCESS in 12m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8a67ceb8d1244246ba3d16039bb1e1b2 : SUCCESS in 15m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/01eca7b4c5d94606b496b259efe4bb37 : SUCCESS in 18m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6415717b1d994b23a5006687cdb7f3f2 : SUCCESS in 7m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/90a6bb8ce33f4e808681aa644ef82abf : SUCCESS in 18m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5bdc5d1244424cdf8f06fd84ff5f0bcf : SUCCESS in 1h 28m 52s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f6528316be304f47849ec9abbeb57556 : SUCCESS in 1h 37m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46f03d919f5b40ceb3cc15647a9a88e5 : SUCCESS in 1h 50m 59s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/cd11b3c2671541f2a087d35a3b2641fa : SUCCESS in 2h 05m 02s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/67df41b079394bbf978675e98d87e624 : FAILURE in 11m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9b9bdfea45a44769a13b33d9b8a5e8e5 : FAILURE in 55m 13s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/37175c8049a44254a491d1f6c73c8698 : FAILURE in 43m 40s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/120b9255234146d8ae182f7b9e3cdbda : SUCCESS in 36m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9275c3b2e0144a72a917d94e85d1ca70 : SUCCESS in 1h 36m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/38e8b4648c3147aa9898a539da745d98 : SUCCESS in 33m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f3b0ac0ec7ab4d609e0864956ccfbe45 : SUCCESS in 42m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/36b5892e98ec4d74a2203ddeb12c6c96 : FAILURE in 14m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ce6b845372801b282d935c17658fd1962fa8449a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-19 17:47:16.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"c4f9dbce5f99ebb93221571c7902eec03716d263","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-19 17:53:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"148c5dbe3fb232e9c6f422aae48acbec0630ec6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 17:54:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7a3974198e1b17d1e91291763e98b46b3019fb96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 20:27:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2dbdfc7c8ae41e1ad29bcfb6bc47665 : SUCCESS in 19m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ae5d28347d548008e70c73bc54762ae : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8036eecec14426ba1c58381aaabf12e : SUCCESS in 16m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2de50a37d8cc4cb6ad8879e7f95e4057 : SUCCESS in 16m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71d6df46c4d24e5f850b24af8ae754f0 : SUCCESS in 11m 47s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8d7cb7d49b814c5ca1ff1985d2f1b535 : SUCCESS in 7m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2fcada7c7f1844819562f480e701c53c : SUCCESS in 19m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/32b7d8c0c6e04779b9230993b61d578c : SUCCESS in 2h 00m 27s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c1903dc944f14fcfa45cef46e50703ba : SUCCESS in 1h 48m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79523f05098a4d84b57285c974dad58f : SUCCESS in 1h 47m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9f2444dcc25e271ad4aad814e3d456362d2c0144","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 20:27:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7d4be689f3924eab4195761189deac06b8bebad3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 20:29:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f088aa50210f4c38b91126b7155983b5 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bb91c6ca759c2cb2ea6d99c4ade7cf025b59d0a1","revisions":{"bb91c6ca759c2cb2ea6d99c4ade7cf025b59d0a1":{"kind":"REWORK","_number":1,"created":"2021-05-17 07:42:27.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/73/791673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/791673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/791673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/791673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/791673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/791673/1"}}},"commit":{"parents":[{"commit":"db24d7bb5187676c0501e722b7eaa5bd9161829e","subject":"Fix weighted round-robin about UDP listener","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db24d7bb5187676c0501e722b7eaa5bd9161829e"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-01 12:11:07.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-05-17 07:42:16.000000000","tz":120},"subject":"Fix comment for the ca_certificates_file opt","message":"Fix comment for the ca_certificates_file opt\n\nThe directive [certificates]/ca_certificates_file currently has a\nconfusing comment. This tries to fix it and make it more easy for\nOctavia operators to configure the directive.\n\nChange-Id: I99ce408ec886820c056b69696b26be9521740f1c\n(cherry picked from commit ee0da827b163c15796bf2f19b0785bc0259d5128)\n(cherry picked from commit 20db85fd5236a4daed7ab29931fed91da2295ab0)\n(cherry picked from commit d69d06660633da7370604a45f43e0014d023f9d9)\n(cherry picked from commit a8977b01ad85d06039956744895262629584f425)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb91c6ca759c2cb2ea6d99c4ade7cf025b59d0a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb91c6ca759c2cb2ea6d99c4ade7cf025b59d0a1"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
