)]}'
{"id":"openstack%2Foctavia~792911","triplet_id":"openstack%2Foctavia~master~I616b5442a01275b26cfcafd262560cdaba65ad4e","project":"openstack/octavia","branch":"master","topic":"story/2008920","hashtags":[],"change_id":"I616b5442a01275b26cfcafd262560cdaba65ad4e","subject":"fix an issue where batch update member will lead to in_use_member error","status":"ABANDONED","created":"2021-05-25 09:05:52.000000000","updated":"2021-08-23 07:46:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"28d4eb3d34d51612e89dd4e81829b39afc808c2c","_number":792911,"virtual_id_number":792911,"owner":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-25 11:41:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-25 11:41:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"540cfafd4438cf244b8180da5d8456ed2e1c1a5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2021-05-25 09:05:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2e607ba82bdbe459ae7f1a82e3b98ff759a2de6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 11:41:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9218712ac0b4a2ea66521f9d6e1b4ff : SUCCESS in 10m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/301a640e822e4a22bdf9dd6b33ffad8b : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19c56fe471174b2594ddab6899310a97 : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0319ddd93b814a829b97d5db5f141349 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/282a4211f0774d5d95e2fb1d1b4b1ded : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e99c8a8503b74709b237cb069a689908 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4f73bda9bba4053924933c14ccc8ed7 : SUCCESS in 12m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/60aadd30178443da8b420afbe56f3f82 : SUCCESS in 7m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3d0b7d262c804a35815d1f2f549423c0 : SUCCESS in 8m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ecc69a5014ac42c28b468894770c7ecd : SUCCESS in 5m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fff0f38ed278405fb5906994b1f2ab66 : SUCCESS in 8m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c63e41a716f54dfdb2e997f3fea3d8a0 : SUCCESS in 1h 06m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6e6811e03ea44ce492f3556152d8d79b : FAILURE in 2h 27m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/29b4191daddf406fa81f515ea700c79a : SUCCESS in 36m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d843eada73a41f2b2bb0e98aead8325 : SUCCESS in 38m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/13478f68a6824bfab91640912a103d31 : SUCCESS in 49m 32s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ab1ca0eba760450f957090c80371708c : FAILURE in 2h 33m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ae375640e78947cd9012c81afb707deb : FAILURE in 18m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1f747b32eae84268b0448b8663e43a6a : FAILURE in 2h 19m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e020a934962b41dea37dd1feab2c90c3 : TIMED_OUT in 2h 33m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/835090f80407498696ea1b7a78ec9bb2 : TIMED_OUT in 2h 33m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b9369478948f48ee9833aa62ee0af39a : FAILURE in 8m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc712238fa49f9f3f6e0849fb8e9812d8f8f273f","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2021-06-09 05:36:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1ef651f6fd976d868bbe18ed599ab9cc3ab909e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 07:44:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22dc0811fbf64c7c8100fd292220a161 : SUCCESS in 9m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb05484d52d94f8c9577b486b8adcd8a : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6725b1c43dc74db29a93c1f1a0da319b : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/968e4cf269234ea194c6cd32c3b7aac8 : SUCCESS in 8m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46235db641834e80a1b106c7f0d0b6e8 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa1200dfc3334786b92c65bac7ded543 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a66dd8e779a4419b70fb58ef979c6a9 : SUCCESS in 11m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/44afdd02127f4a7aa8868f0d6ab9a57d : SUCCESS in 5m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/71a6b9ebf9a944fd898893e58e753b3d : SUCCESS in 8m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/53a80e5001784f499fc9d4995c3419f3 : SUCCESS in 5m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bb2e5e6297af4d3ca24cf65cd28b2d42 : SUCCESS in 7m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ce6ffdb6539840459478246b14e7affa : SUCCESS in 1h 34m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2bb8c940ba72419d8141c845ccb655e0 : FAILURE in 1h 41m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/222cde3395b14391aede264b9a635a4a : SUCCESS in 42m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b6e64bf6c4304e2a9e4c3b002627c862 : SUCCESS in 42m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/40f38ed6440a4008afddf3da60b6b0fe : SUCCESS in 56m 18s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f22c996942114c58af492d5df704899f : SUCCESS in 1h 28m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/046e727d9d7044469c76da95b197ad51 : FAILURE in 16m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ae75fbf77d6d4c94943bf781c9c64d1f : SUCCESS in 2h 03m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/101b41db8b154237861e431d3bc7a539 : SUCCESS in 1h 30m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/344fa36fa44b47cd8d052447077852a2 : SUCCESS in 2h 06m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3236f06d9ad6486bbcf55af7cec55c2e : FAILURE in 11m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"28d4eb3d34d51612e89dd4e81829b39afc808c2c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2021-08-23 07:46:16.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"787ce387f8eee695ea1124b4524af4b05358a448","revisions":{"787ce387f8eee695ea1124b4524af4b05358a448":{"kind":"REWORK","_number":1,"created":"2021-05-25 09:05:52.000000000","uploader":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"ref":"refs/changes/11/792911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/792911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/792911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/792911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/792911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/792911/1"}}},"commit":{"parents":[{"commit":"0a5e7ef91a29896793f2b527c5f5d0c6886e3c1b","subject":"Drop Bionic support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a5e7ef91a29896793f2b527c5f5d0c6886e3c1b"}]}],"author":{"name":"chimeng","email":"chisimon98k@gmail.com","date":"2021-05-25 08:47:23.000000000","tz":480},"committer":{"name":"chimeng","email":"chisimon98k@gmail.com","date":"2021-05-25 09:05:48.000000000","tz":480},"subject":"fix an issue where batch update member will lead to in_use_member error","message":"fix an issue where batch update member will lead to in_use_member error\n\nAfter calling the batch update members API regarding the usage amount of Members does not correspond to the actual Members\n\nStory: 2008920\nTask: 42513\nChange-Id: I616b5442a01275b26cfcafd262560cdaba65ad4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/787ce387f8eee695ea1124b4524af4b05358a448"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/787ce387f8eee695ea1124b4524af4b05358a448"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
