)]}'
{"id":"openstack%2Foctavia~795109","triplet_id":"openstack%2Foctavia~master~I8f0d3ba1d083c7e37e7e26617826400a538a0890","project":"openstack/octavia","branch":"master","topic":"interfaces","hashtags":[],"change_id":"I8f0d3ba1d083c7e37e7e26617826400a538a0890","subject":"Fix using host_routes in VIP subnet with amphorav2","status":"MERGED","created":"2021-06-07 13:37:04.000000000","updated":"2021-06-28 11:06:11.000000000","submitted":"2021-06-28 11:04:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795109-interfaces","meta_rev_id":"19048ea742a4864899b3c77be71762be1e0b80dc","_number":795109,"virtual_id_number":795109,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-28 11:04:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-06-28 11:06:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-06-28 09:03:50.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-06-28 11:06:11.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-06-28 09:03:50.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-06-28 09:03:50.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-06-25 10:35:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-07 15:26:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-07 17:06:58.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-06-08 11:49:50.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-06-28 09:03:50.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"95673ebb1d64c5f3289592bfad6ab8be03a81525","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-07 13:37:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcb283f626d68841e9ef56c1f530bca7897c2040","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 15:26:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c1498786fcf440f9a677168e9d174ee : SUCCESS in 10m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70108ddaff964e3f82773aead3b0f8e4 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c19356e7737a4df49b02417cdd9c37b0 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d95c175010d49f3a5adb344ef0807cf : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6e32448df5c4341879f7e4e0ea9c2a3 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/465574d3744f4e959d3c63ffb091aae2 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49d0e0af35c4491f93be173b4306adc2 : SUCCESS in 13m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da69b3fb56984ca3b2ca5caeebfdfe49 : SUCCESS in 6m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cd21fa79c91d4329b0dd7e2810260445 : SUCCESS in 7m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0cdebb2dbe4c49b1a52f4d045730fd1f : SUCCESS in 9m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cfe6b65fb4a747c2bf92db48eecfbe86 : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/17b436881e99442f8357649ba8689a37 : SUCCESS in 8m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c0210f91639a4381977d58cac4d1e49b : SUCCESS in 1h 41m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28e5a2352dbf4e0ab6c8fd5782fa5940 : SUCCESS in 1h 20m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/07ec57d3e03f4e25a768faa3a55a4dd8 : SUCCESS in 39m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9630fb3651df48c982944ec8fb9ea0b8 : SUCCESS in 34m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aa9edf1986374b05a12c924e63183f43 : SUCCESS in 53m 50s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2b22724dd2d84e15a84060aa21d31ab7 : SUCCESS in 1h 21m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f03e2853d37a446cb76cd223e5ae813c : FAILURE in 18m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f1357ce0b0ce4adeb97bd43e31657db9 : SUCCESS in 1h 31m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9715abac4d9144f9968212a3bd0afdf4 : SUCCESS in 1h 16m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a1e8a3870dcf494a9b28e3e1982c9c0f : SUCCESS in 1h 36m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5478759709b3459e8bd85629b82b164c : FAILURE in 10m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"943e41d6c9d08f70e10f050dd5e5653cef2bb50b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-07 17:06:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d033b01ff9a52eaecfdf989b7cbce241f383115b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-08 11:49:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4f4e13d6c73fc610c4ad38c6fb7b4e533b05e61e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-24 15:51:13.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0860eda4f923d7cc1b8d9892788f4105d4eb2744","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 18:04:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a20e8e43e53a447b9feacd1ebfd1931f : SUCCESS in 11m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/938ef226498e4a268f7bf346b7fb95ce : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90f8989f25ed43e3bebc1888efc1f4a9 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b13262a2506f4a8c9dfb4f6c65bad966 : SUCCESS in 8m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/831d9541ba6e4e4a8ba60fd98e258b9b : SUCCESS in 8m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93f68e6af60544a29a465100d22be1cb : SUCCESS in 9m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26e564ca297b4e26a6a20f6344fec19e : SUCCESS in 13m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b6613dcb6f645439438b3236fac5956 : SUCCESS in 6m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b26177c59c8f43dd897a7e87a0395480 : SUCCESS in 7m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e2c089c0b9784ba3974ee67e8e874ad5 : SUCCESS in 10m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5d3c76b685614763ba2fbbd27f5b03ae : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b08f1a4825f4482f840f70a49e495d16 : SUCCESS in 11m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/486e6b2b0a4c4aaa9b099b5f4a52b76f : SUCCESS in 1h 53m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7e2943980e604e0491d1cf211cf9a7f2 : SUCCESS in 2h 06m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d33c174424d2422e9a9768028c09fc5f : FAILURE in 37m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc973eb2e5a7400c9d145bc4794aff2b : SUCCESS in 41m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ce45ceeb5ca7483eb243257fdd09455d : SUCCESS in 1h 03m 24s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d4c0814c4ee3431dbf69b9b1292b3e79 : SUCCESS in 1h 59m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4d7c5e13ffc54604b04d8f3d5f56537f : FAILURE in 17m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9f96addec2db4d02a1a5915d7bdb2907 : SUCCESS in 1h 57m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e01b61f97d1e4c28a6681ad0456bff23 : SUCCESS in 2h 10m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/838fa0144af140059def1053795404a3 : SUCCESS in 2h 03m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/118cac5748534be1bd44b8363097fe25 : FAILURE in 10m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6082c0cffc9071952187f1397d2ad951c4f93405","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-24 19:13:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"27db6ca5ae4ba2961b2cd812b52411a226a399f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 21:29:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7635219ccd7647c2a09a068569b277ba : SUCCESS in 12m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/336e01353b6a4214b13f2b031f9fb7a8 : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bee3e5edd8174b399c321e7b7ca59587 : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b5d2c6ecb2a4365ad6aabbca5ed0c2d : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf9f3c1303c7417484c048916105e7f2 : SUCCESS in 7m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8daaba816cf341a1adcf5fa34ca7c815 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beb74d854e6f4cb385138b1a3488e7d2 : SUCCESS in 15m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f7f50356005404cabc26929902e54a6 : SUCCESS in 7m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/36d32f6f0e434bdbab59d2bc403bfecb : SUCCESS in 10m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cf6d5cf3edef4822a9bec48c99abe40e : SUCCESS in 11m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ed80c19f59b2494da29ef675d8647ada : SUCCESS in 7m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/06f2d7d68a9542399aa146882798bc1d : SUCCESS in 10m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9512b171c6704f4abda60cdbabe8b1e9 : SUCCESS in 2h 14m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0df222ec8d614e05978bd45b90972acb : SUCCESS in 1h 23m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b3691d40093c4955b6466ff647166477 : SUCCESS in 41m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/47cf1a3313df44bd833d8d0e312f16b8 : SUCCESS in 37m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1751a0de20eb4b8bb55a91b24f725cbd : SUCCESS in 50m 26s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dd53c5efa352427bba9a488f873eb317 : FAILURE in 1h 36m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bbaa4539143f4d20b00268ffad931b76 : FAILURE in 17m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/528e5fd625b74db2a78707f980ee14e5 : FAILURE in 1h 53m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/331c24b250174cc2abf602b419fc8333 : SUCCESS in 1h 29m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/42d0b4ab8c3b4c9cb536ac941d4fa368 : SUCCESS in 1h 25m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3bf992307e02421bbc3a1dcd21a3b191 : FAILURE in 11m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8ad02a322ba4d2910543c8de0f5b620649970d46","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-25 10:35:23.000000000","message":"Patch Set 2: Backport-Candidate+2","accounts_in_message":[],"_revision_number":2},{"id":"2f39c286d6f47b00dab21edc4bdcfd2bb744ee8e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-06-28 09:03:50.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2aa1c42d04cbb5e3e29c01775e53f83683b587c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 09:04:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"404d3f52b5e468ead61621873acc4b73ff944f22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 11:04:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82e2effab8ea4e498335e1aeccca5356 : SUCCESS in 14m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83c4dea9cfaa4855bed39797ffaeb04f : SUCCESS in 8m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95736ffc0e934a759897da21725eebd6 : SUCCESS in 10m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0f25a3900f24f5696c20d3f251deae8 : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e895d65a08f4260b672b7abb95ae2eb : SUCCESS in 23m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70327f31e649441ca25b8cdfa0ceb3a3 : SUCCESS in 8m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/900b9aa867e5470fbaf63332191b7c74 : SUCCESS in 9m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/acdb500c5ebb4217ad1a98a2d30a6014 : SUCCESS in 13m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/97cf17f8f4c141fd8e83a6327eca744c : SUCCESS in 1h 54m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5a24cffd28046a8a188d8ad17bbf78a : SUCCESS in 1h 46m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/205d3ed74b8c4a8fbe59300d3999f840 : SUCCESS in 43m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a152645e7d84f95a859b70deb517953 : SUCCESS in 54m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a05a98dd44694aee83f0947e62748db2 : SUCCESS in 49m 12s","accounts_in_message":[],"_revision_number":2},{"id":"ed12ccb57a44a66b7709d51609dedd7a08a5245d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 11:04:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6a4b3a702f42be650ca9d6f31af4e7af907b89d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 11:06:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6317d923d6824af7bc0fa1ea773f7c1d : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4df8b9f49eed41cb99a394b906b3abda : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"76606b4c4dc9595a015afb4c997f5d02ac7a7343","revisions":{"9db74aa4a3dfec9225bbaffe29026f93a6bba4c4":{"kind":"REWORK","_number":1,"created":"2021-06-07 13:37:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/09/795109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/795109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/795109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/795109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/795109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/795109/1"}}},"commit":{"parents":[{"commit":"e1adb335c788dbd902bfd8a00f748da4abb8b06b","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e1adb335c788dbd902bfd8a00f748da4abb8b06b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-06 06:00:36.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-07 13:36:44.000000000","tz":120},"subject":"Fix using host_routes in VIP subnet with amphorav2","message":"Fix using host_routes in VIP subnet with amphorav2\n\nEnsure that amphorae_network_config parameter is not modified in\nAmphoraPostVIPPlug. Modifying it would break amphorav2 persistence by\nintroducing non-serializable elements in the subnet dict\n\nStory: 2008955\nTask: 42580\n\nChange-Id: I8f0d3ba1d083c7e37e7e26617826400a538a0890\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9db74aa4a3dfec9225bbaffe29026f93a6bba4c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9db74aa4a3dfec9225bbaffe29026f93a6bba4c4"}]},"branch":"refs/heads/master"},"76606b4c4dc9595a015afb4c997f5d02ac7a7343":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-06-24 15:51:13.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/09/795109/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/795109/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/795109/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/795109/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/795109/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/795109/2"}}},"commit":{"parents":[{"commit":"1d77e93f2889627cb87f1aed3259b83e91e83239","subject":"Merge \"Fix new errors with SQLAlchemy 1.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1d77e93f2889627cb87f1aed3259b83e91e83239"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-06 06:00:36.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-24 14:43:41.000000000","tz":120},"subject":"Fix using host_routes in VIP subnet with amphorav2","message":"Fix using host_routes in VIP subnet with amphorav2\n\nEnsure that amphorae_network_config parameter is not modified in\nAmphoraPostVIPPlug. Modifying it would break amphorav2 persistence by\nintroducing non-serializable elements in the subnet dict\n\nStory: 2008955\nTask: 42580\n\nChange-Id: I8f0d3ba1d083c7e37e7e26617826400a538a0890\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76606b4c4dc9595a015afb4c997f5d02ac7a7343"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76606b4c4dc9595a015afb4c997f5d02ac7a7343"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
