)]}'
{"id":"openstack%2Foctavia~795356","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I1281e74a398d03eac8db7a468e71ea4889dd8229","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I1281e74a398d03eac8db7a468e71ea4889dd8229","subject":"Explicitely set nodeset to Bionic-based","status":"MERGED","created":"2021-06-08 14:48:41.000000000","updated":"2021-06-17 19:15:39.000000000","submitted":"2021-06-17 19:13:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795356","meta_rev_id":"a3a1408b541c3a95b4fce1099d1442558e2fc6f6","_number":795356,"virtual_id_number":795356,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 19:13:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-06-09 07:50:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-06-09 19:56:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-06-17 17:27:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2021-06-17 17:27:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-08 17:35:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-09 07:50:10.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-06-09 19:56:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-06-17 08:05:42.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2021-06-17 08:05:42.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-06-17 08:05:42.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"1fcc4eefd5afda7d6f9acb85c3859c5d6b552493","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-06-08 14:48:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bd5dc58796fdc806f268a8c0c524354c175d543","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 17:35:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9c3eb5624ac4b75b92e518c0437a880 : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26b9d72f368c49b2a8891dce96f26ee3 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2a6b7a4fee84ede8e55c7d5e2d990f8 : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a971455513b14f0eb1a29a155690c794 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab1839bc57d64a05948b03d9cd3b3236 : SUCCESS in 13m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/756b83e489194099be02884090766941 : SUCCESS in 7m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a8d3de4ddc3a4fc7885fc28ca079a816 : SUCCESS in 8m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eb1eed99d846499ca7f63f021647518f : SUCCESS in 7m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eb41ab68ae8f4d04a3f648277eb882f8 : SUCCESS in 1h 36m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/07d35d83dac64d1eb4676b5524e26b05 : SUCCESS in 1h 44m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a912f599c98d42f68f87771ac6e76b53 : SUCCESS in 42m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/26347eb501724d22896752b071fb1670 : SUCCESS in 43m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f15e42fe6a254042bbc1636f12651967 : SUCCESS in 44m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/367d1e00bfb84616a6a32753693d2346 : SUCCESS in 1h 11m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9d4d679fc9c4409097132b4bf4f4e480 : SUCCESS in 1h 33m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/25ff80756f404321a91fc2bd9cbb555f : RETRY_LIMIT in 8m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0650ece9ffba44f08c80af554004e8c1 : SUCCESS in 1h 58m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da95821d6bb5044ddda118376edf37d89af3d63f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-09 07:50:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7a5925da19afa3487a190f2dbfd3bad727c3323c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-09 19:56:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b63e05552070f74309c462704996b7141c0b727f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-17 17:27:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM, thanks!","accounts_in_message":[],"_revision_number":1},{"id":"3738a394ffc23ea46db95f8a3215d59b84e31665","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 17:28:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"50caa67adc49acebe36f44cbcab2852deb7309ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 19:13:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/809995ced49d4c06b54ca0a3baeed5f0 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a2e79b0422f469197b751e9b4c2c161 : SUCCESS in 6m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/11c1f84702f14b60b6cb0691f845115b : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a9930fb83a0415aa87acf2ab4acb225 : SUCCESS in 12m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5bf98581c9514250aea6429ee84510e8 : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d3031c67e5664722800fd9b19ce03439 : SUCCESS in 1h 27m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bb912d2c85fe44cdbdfbae637ba0e96c : SUCCESS in 1h 44m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1a3b30a2345a45999f98dd91d3c02ec0 : SUCCESS in 37m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6389d6735ebe408a8f4f5af36aed6e3c : SUCCESS in 50m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3a4b8c93e4514b3e817f95db2290b4e7 : SUCCESS in 41m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b73cb987f63441a5992060736f8e9241 : SUCCESS in 1h 00m 17s","accounts_in_message":[],"_revision_number":1},{"id":"7c3f18d8d43e2c79a01b49ea7a9f13bba0ba9945","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 19:13:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a3a1408b541c3a95b4fce1099d1442558e2fc6f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 19:15:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19de43ec3e204beb93380f2b53c5deaf : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b6d25ca57bc383e4a110da84b8c325eb5ec20e2","revisions":{"5b6d25ca57bc383e4a110da84b8c325eb5ec20e2":{"kind":"REWORK","_number":1,"created":"2021-06-08 14:48:41.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/56/795356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/795356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/795356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/795356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/795356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/795356/1"}}},"commit":{"parents":[{"commit":"7a4e04c8099fcdeaab043ecef743bcf0e71e6164","subject":"Merge \"Fix LB failover for amphorav2: set security group\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a4e04c8099fcdeaab043ecef743bcf0e71e6164"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-05-25 12:44:42.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-06-08 14:45:28.000000000","tz":120},"subject":"Explicitely set nodeset to Bionic-based","message":"Explicitely set nodeset to Bionic-based\n\nThe default nodeset has been updated to Ubuntu Focal in [1]. Ussuri CI\njobs need to be explicitely set to Bionic-based nodeset or else Devstack\nwill fail to deploy with the following message:\n\n  \"WARNING: this script has not been tested on focal\"\n\n[1] https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/788954\n\nChange-Id: I1281e74a398d03eac8db7a468e71ea4889dd8229\n(cherry picked from commit 46a5344e2bab2d2ccd6dad5204fcb8ab421b0e06)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b6d25ca57bc383e4a110da84b8c325eb5ec20e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b6d25ca57bc383e4a110da84b8c325eb5ec20e2"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
