)]}'
{"id":"openstack%2Foctavia~796994","triplet_id":"openstack%2Foctavia~master~I64b5c70fca7f1cacb2ef10c2b61f28b9000c7504","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I64b5c70fca7f1cacb2ef10c2b61f28b9000c7504","subject":"Fix new errors with SQLAlchemy 1.4","status":"MERGED","created":"2021-06-18 06:38:00.000000000","updated":"2021-06-21 08:38:25.000000000","submitted":"2021-06-21 08:36:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796994","meta_rev_id":"33596618ab551181ea89e35f36ee9b0bed4a789e","_number":796994,"virtual_id_number":796994,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-21 08:36:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-06-21 06:42:21.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-06-18 16:51:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-18 15:23:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2021-06-21 06:42:21.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"disliked":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-1,"date":"2021-06-21 06:42:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":-1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-18 08:45:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-18 15:23:41.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-06-18 16:51:49.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-06-21 06:42:21.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"fb0ba8aa142a68e01cf805244644501b95718e07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-18 06:38:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"719da06cec7d247e59f45e58f853aa3711cf62e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 08:45:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e45d8c5c3b9f491582665070660d1001 : SUCCESS in 14m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1dea16c7df354fc38deaf265c13a571b : FAILURE in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a07ce8335486407b84bd33ad7f053ed5 : SUCCESS in 8m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e427c3507784fc4a5bf434ed3a53061 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60cc5f09ab2d4b9aa54b1241b6423d63 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09bf55fdc51141e39dcd3c77f5b4b7a3 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d50386bf678442ea4cdb246b58eb85d : SUCCESS in 14m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/835c671a6a694050a3a63a1ed8960a89 : SUCCESS in 7m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d444836d31d84c6c9c7f5b521fe66c88 : SUCCESS in 10m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dfc36f8043ae4ea196c638a9136f5465 : SUCCESS in 7m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aa10e9234d8e457f80d7daa572cb7361 : SUCCESS in 9m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f16da3bb94c04bd3b010e7c27ca4bc93 : SUCCESS in 1h 58m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/132e709dc6f84e078d9b3cbdde3d1287 : FAILURE in 1h 38m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/93c6c2df0fc34a0da54e1f22723cc92b : SUCCESS in 40m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9efba1e66d274f1b83601ab716bb2217 : SUCCESS in 44m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/345f29b83faa4fd3a13635c5c5546d59 : SUCCESS in 49m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/678cf01063d64f4987efa5a01243f5c8 : SUCCESS in 1h 38m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bcd19da861a64e9fb75c58e42d8b7d09 : FAILURE in 22m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b5fe701053424b82bac7b0cc083990bf : SUCCESS in 1h 30m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a84bf799e8d94c019732006e7512b433 : SUCCESS in 1h 41m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/23e5d1a4684a44f5884da56c6d53c475 : SUCCESS in 1h 28m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4ff6c2208a9340a1b3409a384e2a9dc4 : FAILURE in 15m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99fb130e55d98e1b2db5d70fe148b131228fb435","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-18 09:29:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"012b4f6a7e9040d7e16e6ee41871bceca19ef572","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 11:44:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9701a9b3507543f78593d4d6d0b472c5 : SUCCESS in 11m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f4be2292d9c4747b0d7ab02a9b3cb00 : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db570115ce3844b5a461e34e78b5fe69 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad4114c3dfba450ea219fd0947f2d5da : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b7534f67b2641248ff549df69fd95c7 : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fa3274acdc34c05b477046dc12f6268 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48f0be599a0a43d68a1cb8c7dd5b5481 : SUCCESS in 13m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f1c2d61fd5104d69803999a2dfd2f73b : SUCCESS in 6m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3fd9952a69ef4b6d8b1208d992901e0a : SUCCESS in 10m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/72e777193adb4f2893e2d7d99c94383f : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/448aba6219cf46f4bf63e6a5b68b940e : SUCCESS in 8m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0aee81cf9b384be08f7f997f1c4ccff4 : SUCCESS in 2h 07m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5f73eec90b149a988cd2b2238b4c6f0 : SUCCESS in 1h 35m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/62a580a3ae25454895051608d2eb9676 : SUCCESS in 38m 38s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8923c6c2546747e3a7c2594a80ad8e60 : SUCCESS in 41m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7a673c3d372b4e8181143ac78b55f16d : SUCCESS in 56m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f4680c2f7b1d404a80e6321d48c4ea60 : FAILURE in 1h 33m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/eead487e1ffa49d898ac331547b22c7f : FAILURE in 18m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e87bebe9d1964034b682aba9d012596b : SUCCESS in 1h 47m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d00b556a16ef474798458ca458f93518 : SUCCESS in 1h 39m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5df6fbedff1348c69100ef3e251417e3 : SUCCESS in 1h 35m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d9ba0b1717604f75b9a31ea74531ea55 : FAILURE in 9m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2f0ba2326b7ae2d69e614b9b5a70f66a75415b64","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-18 15:23:41.000000000","message":"Patch Set 2: Code-Review+2\n\nGreat detective work Greg, thanks!","accounts_in_message":[],"_revision_number":2},{"id":"80772f0f4546987605633386950d27cb185e4a03","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-18 16:51:49.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"51405543ddc15ee1a1bf5b602fc4339ad684a5d3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-06-21 06:42:21.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+2 Workflow+1\n\nThanks Greg!\nI don\u0027t think this needs to be backported as latest released version (Wallaby) has SQLAlchemy upper constraint to version 1.3.23. Just BC-1 in case someone sees it differently and/or requires this backported.\n\nhttps://github.com/openstack/requirements/blob/aee7a4b34ee91065581ae20ba81d30920bb18769/upper-constraints.txt#L151","accounts_in_message":[],"_revision_number":2},{"id":"092d4c9465f85d6dcf8c03285ebb80ddb67ddda7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 06:42:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a237249ba173f3f5e2dd80269289f7a52bedc865","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 08:36:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db1a88d9ad484aff8a473199a855759b : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12a6960b73994387b21146ef61cfa6d3 : SUCCESS in 8m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ef8bc6844b2413b9f3d9ac0f8e5cfcf : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80e0346b9aaf42b99860d5df8066c2dc : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c68c8de946444c969f32aa07fc15eaae : SUCCESS in 16m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4c281df3f4894006b963dd08eeb405f4 : SUCCESS in 7m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/27f5f3a0628e439d9d38147d86fa0c7f : SUCCESS in 10m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ff27bdeec3974e39988bdaf53ef2f14e : SUCCESS in 1h 49m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/80d6081ab4064ca4bd4c979ddd86743b : SUCCESS in 1h 43m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1e6e6f2e14ba4878837f9fa9433f9c40 : SUCCESS in 38m 33s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fbc34b44ca1b41b483ab59a34fc96d9d : SUCCESS in 38m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/eb972ab893b344caba175726b694e0d3 : SUCCESS in 53m 13s","accounts_in_message":[],"_revision_number":2},{"id":"afa5bccddcd292a28ac0bd144e3a8e5ab79dd49e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 08:36:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"33596618ab551181ea89e35f36ee9b0bed4a789e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 08:38:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10db35eed0c5456f8fb775ff70ae2072 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c58a2c7c65feb7a9aa74c1f116865e14da511614","revisions":{"95020ca1ecfd04d722ae8542dce82a35ecaaea11":{"kind":"REWORK","_number":1,"created":"2021-06-18 06:38:00.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/94/796994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/796994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/796994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/796994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/796994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/796994/1"}}},"commit":{"parents":[{"commit":"e1adb335c788dbd902bfd8a00f748da4abb8b06b","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e1adb335c788dbd902bfd8a00f748da4abb8b06b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-18 06:05:31.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-18 06:37:18.000000000","tz":120},"subject":"Fix new errors and warnings with SQLAlchemy 1.4","message":"Fix new errors and warnings with SQLAlchemy 1.4\n\nFixed in Repositories.get_amphora_stats:\n  TypeError: unsupported operand type(s) for +:\n  \u0027ImmutableColumnCollection\u0027 and \u0027list\u0027\n\nFixed in many places:\n  KeyError: \"Deferred loader for attribute \u0027id\u0027 failed to populate\n  correctly\"\n\n  Modifying an object without using a transaction created some issues\n  with the next query. SQLAlchemy flushed and expired the modified\n  object while it still needed to read one attribute. This is a weird\n  behavior in SQLAlchemy\u003e\u003d1.4 (perhaps a bug), but using transactions\n  seems the correct way to update the objects.\n\n  Note: The message in the SQLAlchemy exception is totally bogus.\n\nFixed warnings with the _tags relationships:\n  relationship \u0027LoadBalancer._tags\u0027 will copy column load_balancer.id to\n  column tags.resource_id, which conflicts with relationship(s):\n  \u0027HealthMonitor._tags\u0027 [...] To silence this warning, add the parameter\n  \u0027overlaps\u003d\"_tags,_tags,_tags\"\u0027 to the \u0027LoadBalancer._tags\u0027\n  relationship.\n\nChange-Id: I64b5c70fca7f1cacb2ef10c2b61f28b9000c7504\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95020ca1ecfd04d722ae8542dce82a35ecaaea11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95020ca1ecfd04d722ae8542dce82a35ecaaea11"}]},"branch":"refs/heads/master"},"c58a2c7c65feb7a9aa74c1f116865e14da511614":{"kind":"REWORK","_number":2,"created":"2021-06-18 09:29:38.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/94/796994/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/796994/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/796994/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/796994/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/796994/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/796994/2"}}},"commit":{"parents":[{"commit":"e1adb335c788dbd902bfd8a00f748da4abb8b06b","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e1adb335c788dbd902bfd8a00f748da4abb8b06b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-18 06:05:31.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-18 09:25:48.000000000","tz":120},"subject":"Fix new errors with SQLAlchemy 1.4","message":"Fix new errors with SQLAlchemy 1.4\n\nFixed in Repositories.get_amphora_stats:\n  TypeError: unsupported operand type(s) for +:\n  \u0027ImmutableColumnCollection\u0027 and \u0027list\u0027\n\nFixed in many places:\n  KeyError: \"Deferred loader for attribute \u0027id\u0027 failed to populate\n  correctly\"\n\n  Modifying an object without using a transaction created some issues\n  with the next query. SQLAlchemy flushed and expired the modified\n  object while it still needed to read one attribute. This is a weird\n  behavior in SQLAlchemy\u003e\u003d1.4 (perhaps a bug), but using transactions\n  seems the correct way to update the objects.\n\n  Note: The message in the SQLAlchemy exception is totally bogus.\n\nChange-Id: I64b5c70fca7f1cacb2ef10c2b61f28b9000c7504\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c58a2c7c65feb7a9aa74c1f116865e14da511614"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c58a2c7c65feb7a9aa74c1f116865e14da511614"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
