)]}'
{"id":"openstack%2Foctavia~797882","triplet_id":"openstack%2Foctavia~master~I15997acaf12258ec483286dad676efdea8963611","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I15997acaf12258ec483286dad676efdea8963611","subject":"Enable lo interface in the amphora-haproxy netns","status":"MERGED","created":"2021-06-24 11:43:00.000000000","updated":"2021-07-01 00:55:21.000000000","submitted":"2021-07-01 00:53:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797882","meta_rev_id":"28038716123739427dff949c3d588a0c35b9aa6b","_number":797882,"virtual_id_number":797882,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-01 00:53:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-06-30 14:38:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2021-06-30 15:15:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2021-06-30 15:15:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-06-25 10:35:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"}],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-24 13:35:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-25 19:15:03.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2021-06-30 14:38:29.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-06-30 15:15:17.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"02bd47054babdac2fb8da3e44cc4aca9ab7c69b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-24 11:43:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6aef16dfbae6242a738d0fea8218e2ad9e0be4d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 13:35:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbcef1dbed124672a1cd1931fb3c6bc7 : SUCCESS in 12m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47fd668d7bde4562ae2baa420a30684b : SUCCESS in 11m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/800e366009f7477b99d9220b3fb77c62 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d3277acf66a4614ba5efe705dbd107e : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de095715777c418bbc23ba944b026c26 : SUCCESS in 7m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7eeca273f96e4d039518832f7920f8ba : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/930620c444114de8bc93fb64a1c7e219 : SUCCESS in 16m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ded2bdcda3840bb9820269172e4fbce : SUCCESS in 5m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5f2e32b3371043378cca4d8588c3249f : SUCCESS in 6m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ebc41a38200b4f7fae1409a1ef7d3f64 : SUCCESS in 10m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/30d5427e63cf48caa672f2a6dda269e3 : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d9c415d0d5984aca9f99968aba9a51ae : SUCCESS in 11m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8daeb02b5498457088b2f412c2d00f62 : SUCCESS in 1h 50m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/45c75f26b8914dfe9e7c7fd5daeb5d19 : SUCCESS in 1h 16m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d3503417227443ad83a6652f97eb9ff5 : SUCCESS in 41m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/48e0e68cc753492b87ab9b72dfe67456 : SUCCESS in 38m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1ddc7d72a7524678a0370f794951e7b0 : SUCCESS in 51m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c3aadf5e8c83452e83e13ee44e674dc1 : SUCCESS in 1h 36m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b5f1e0b5fd29475998558b4d1c095696 : FAILURE in 17m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/426a2158ff58431a93f5b298a10909e3 : FAILURE in 1h 46m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1128a86639c14323b652ef066935230b : SUCCESS in 1h 46m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c7a3f1be97b442d1b46e183aa1d8ae2e : SUCCESS in 1h 46m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/21774c7fd95a408ca05b92acc3035f83 : FAILURE in 11m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b2f1cc940c41ea7af9402dbe3eb0a40260bec18","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-25 10:35:29.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"aa0f200e2fce13655f544ac93f8ffcf4997e62b9","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-30 14:38:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"37f0cec81ac570acc802157e052369be229cea0e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-30 15:15:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"21869de3231565bf2e0ce795c687f583a2171ebc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 15:16:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"becc5ca4bbcc1876cbb2f5652e26391f009b4b07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 16:29:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08b9192249864f50a27162fd68fe013d : CANCELED in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3298f34833f04da6a0feaf34edb7a7a3 : FAILURE in 8m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33b87b30d9c0427dbff6f5aa1e1e358d : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94c0578a78b1459bbdb07ffc92660ea6 : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a48dd8684b2d4009913dcfef0180ceb2 : CANCELED in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40f15b0af5d5499eab3cee6225708fc4 : CANCELED in 5m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4ebb118eedd74145a99f59e021de7d67 : SUCCESS in 7m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/47b57fbd61a44475be8964cd57aa1224 : CANCELED in 8m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/61331b2600784b858839f58f8755bcb4 : CANCELED in 12m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/439b1ea2cc5647ec977c8ba0a3e9a3ee : CANCELED in 11m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e00da01a90a749308a68227ca0f6b1c8 : CANCELED in 10m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d27bed3c91994e8fa724f6820aaf7886 : CANCELED in 10m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/41922f4adca44e3eb45f64e263f224e2 : CANCELED in 11m 05s\n\nWarning:\n  Comments left for invalid file octavia/api/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v1/driver.py\n  Comments left for invalid file octavia/db/repositories.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v2/driver.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/api/v2/controllers/load_balancer.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/api/common/pagination.py\n  Comments left for invalid file octavia/amphorae/drivers/health/heartbeat_udp.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/lifecycle_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/compute_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/lifecycle_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/compute_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/database_tasks.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/network/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/common/policy.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/common/data_models.py\n  Comments left for invalid file octavia/certificates/generator/local.py","accounts_in_message":[],"_revision_number":1},{"id":"8360b9ceb9bfffba21dfd6542cd1adfb6adac526","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-30 20:53:54.000000000","message":"Patch Set 1:\n\nrecheck\n\npylint fix is merged","accounts_in_message":[],"_revision_number":1},{"id":"4d2ed1e50889f93721e8b37119cb15d9fe6c1c98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 22:56:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ae2e4c442934f298ae88ce53c004e6c : SUCCESS in 12m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e40dec1c4f314677ab0a9324c7b3cba2 : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/325963ffb33f427da595143b896c392c : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c70fc40554c24ef38be5bbbd851c2bc4 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0723fc32b13943f19a8d5436d8788beb : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eae86a81f61641e5a3265643b9dbfd5c : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2d240980c744bceae53399997e65f07 : SUCCESS in 11m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0e9d722fa3c4b9e96a5f77873cefd04 : SUCCESS in 5m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/702397b7b2834fb590df13b008379669 : SUCCESS in 6m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c721ca2eb4d94b229f61199d075e5a3c : SUCCESS in 11m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d5b1958197f54bad886aa0ca5bc3514e : SUCCESS in 5m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/258a78ee48ca46488ad5f7a63a5ac2a9 : SUCCESS in 8m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d8f5f3f747f740c7a614aea8b949631c : SUCCESS in 1h 40m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6a471798b16e4e3c8c26d849da4117db : SUCCESS in 2h 01m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b343fde870814bfa91d203fb14a36d12 : SUCCESS in 38m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71ff3a2fcda842c1a622652340b85dd9 : SUCCESS in 36m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bb060e20a3524cbd81dd65d06391791a : SUCCESS in 51m 19s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4dfb2cd9982c4080b7295b50c38f379b : SUCCESS in 1h 21m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/867879978b634f2d9ea4f78970901ce2 : FAILURE in 18m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f0222ceaf4a7466480791ec417fe3039 : SUCCESS in 1h 49m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/72f7e34acda549e9bc2447cc55c50266 : SUCCESS in 1h 43m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2532a68ff8e74904953a75248938b8e0 : SUCCESS in 1h 54m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ceed073298ee43ba9b7f48b7e04b56bf : FAILURE in 10m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1dcba7c3c7526d8739499e6c989c9b3fb5d57a75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 22:57:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec5f83e41e605164f33c5d19559a3f613a3dbffd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 00:53:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62da2d18568f4d2cad84844a6c5916d6 : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ab3e2dbdcc04da3b9df034979eb8c3b : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63c3c523fa6b4918a13ede9eef069366 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cddd0f23f105490c98392399d73478d5 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8b11e39a555476ea59335e46c94bed1 : SUCCESS in 13m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cab383e5cd454d388b9f822fd82532f2 : SUCCESS in 5m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5629c3ead718465b8e8aab6e01f050e3 : SUCCESS in 6m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d011d18d311c45fe8a0503e86528d3e4 : SUCCESS in 9m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/118a525701354861bb99239bcac993e8 : SUCCESS in 1h 51m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0c24e1c2de24425bbf65f70e886cc9a1 : SUCCESS in 1h 22m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0bd303f99d964500b9ef22e94153abb7 : SUCCESS in 43m 33s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/471b06ff21664d8685c94183f4b6ef4f : SUCCESS in 39m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/55640357a17d434484aa99d3eb9a72ac : SUCCESS in 1h 05m 14s","accounts_in_message":[],"_revision_number":1},{"id":"01f3fa395c43ca667037a40b88bf9e0df6eafbd5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 00:53:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"28038716123739427dff949c3d588a0c35b9aa6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 00:55:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e536304265054d9cbb09282b7c577701 : SUCCESS in 1m 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f0df1dc7eea4f2baf9168ebc060acd4 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4443596e298df53b23b7d81d703479aff186ff4f","revisions":{"4443596e298df53b23b7d81d703479aff186ff4f":{"kind":"REWORK","_number":1,"created":"2021-06-24 11:43:00.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/82/797882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/797882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/797882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/797882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/797882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/797882/1"}}},"commit":{"parents":[{"commit":"1d77e93f2889627cb87f1aed3259b83e91e83239","subject":"Merge \"Fix new errors with SQLAlchemy 1.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1d77e93f2889627cb87f1aed3259b83e91e83239"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-24 08:14:03.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-24 11:42:34.000000000","tz":120},"subject":"Enable lo interface in the amphora-haproxy netns","message":"Enable lo interface in the amphora-haproxy netns\n\nConfigure the lo interface in the amphora-haproxy namespace.\nIt fixes issues with the synchronization of haproxy peer/stick tables\nwhen updating a load balancer configuration.\n\nStory 2009005\nTask 42682\n\nChange-Id: I15997acaf12258ec483286dad676efdea8963611\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4443596e298df53b23b7d81d703479aff186ff4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4443596e298df53b23b7d81d703479aff186ff4f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
