)]}'
{"id":"openstack%2Foctavia~798416","triplet_id":"openstack%2Foctavia~master~Ia815461a15bff065fd63e9ce80636e2b736bebdf","project":"openstack/octavia","branch":"master","topic":"update-min-tox","hashtags":[],"change_id":"Ia815461a15bff065fd63e9ce80636e2b736bebdf","subject":"Changed minversion in tox to 3.18.0","status":"MERGED","created":"2021-06-29 06:00:43.000000000","updated":"2021-07-01 11:30:18.000000000","submitted":"2021-07-01 11:28:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798416-update-min-tox","meta_rev_id":"22fa827ad2ab66f9741c6cd2c9332697d6c4c434","_number":798416,"virtual_id_number":798416,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-01 11:28:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-06-30 07:20:26.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-06-29 13:12:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-06-29 23:56:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2021-06-30 07:20:26.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-2,"date":"2021-06-30 07:20:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":-2,"date":"2021-06-29 23:56:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-29 08:13:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-29 13:12:00.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-06-29 23:56:38.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-06-30 07:20:26.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-06-30 14:36:39.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"CC"}],"messages":[{"id":"4418219fa4a88325862efa9bedad3c3de92431d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-06-29 06:00:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e0caaf567d605daca96f30207f2cb168a49e0da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 08:13:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c7633646a0574cdbac936f9d67a4c64d : SUCCESS in 8m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2070d403b694740a441f73371b67aba : SUCCESS in 15m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aac4dc8f9611449e8e1c5e920dfaad03 : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d7af20c1a174b4cbe1a30c8a7914525 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/531520e651a342d0a1558f53d8417c22 : SUCCESS in 8m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/425beccd322c4754993dbd6d00b653d5 : SUCCESS in 7m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02483b6a551b48c1b134264085b1001c : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/443b9ce841654f6c8ed72234b4c969e8 : SUCCESS in 16m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27ce5da4708d4afa8bd74500738f11b5 : SUCCESS in 9m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ee4fd7cf76bf4fbf987acde9e4db23cb : SUCCESS in 8m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cebf49e934d44edbada0ca58506dd917 : SUCCESS in 13m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/58342e0e9d374deda0ec3cb184738d4d : SUCCESS in 7m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/17135c4f1887477aa05d11ea6968d575 : SUCCESS in 11m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b716f7eb6fe54089b1cffc3b3842a523 : SUCCESS in 1h 53m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff7861bfe76a4d79a64aa66237ad6982 : SUCCESS in 1h 22m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b490fed139594a89893edbc3f01c3763 : SUCCESS in 48m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1100de53824843f38fb718fd55fc5c11 : SUCCESS in 50m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9857ea02c47b4dcd95c962351d5d58e3 : SUCCESS in 1h 22m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/79ef561fda1a40f7a81d1b0d7a979431 : FAILURE in 17m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/296c414a7ee24134b86f4282ce2d0508 : SUCCESS in 1h 37m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f868d7596f52448a9a8650eaddbb2a31 : SUCCESS in 1h 28m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a735688a7cc447408c77238f61b7c051 : FAILURE in 1h 42m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/db512687a46243c98f52561ff3d0faca : FAILURE in 11m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1451ffd2c064bb9c0e6403cde55e0de7af346f26","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-29 13:12:00.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks the same as other changes on this topic.","accounts_in_message":[],"_revision_number":1},{"id":"c94474fbb350dea216c2dbc5e7d70511da8aa43c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-29 23:56:38.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"a7b46c8e5c9e4f75931bab838033474d8c8fe3a1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-06-30 07:20:26.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c3f72c0f73338901cb5019946e78deaf2855b00c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 07:20:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ebef575262c2a257b2a98b7005eee885bab9dd82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 07:35:16.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4731f9b1f53a49eca98971f6666ad363 : SUCCESS in 7m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6251b62bcadb4d81b8d3c16a41ef4765 : CANCELED in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87738f6acb9c483aa8d453f6729d8d76 : FAILURE in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66b8cdbd965e4238a14ea1f8db93063f : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9c2a817887e40fbb4517a3424595636 : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84a5183ee647431a89c3f12ec7598593 : CANCELED in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8b6239c352e4784a134793edd43b07f : SUCCESS in 6m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d7f00f0c159e46e4a84cbcfd6b8ab938 : SUCCESS in 7m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f210fc9a511d4725a9843c710e39f94b : SUCCESS in 10m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c75c18d84d10467ba9fe2307f643d80e : CANCELED in 10m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b8f629d1f9894db0a4d4418245a9fb96 : CANCELED in 13m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/394da90224de419dad8c211218f442ea : CANCELED in 12m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/808f9520e6a44ba18b80b4836f5af4ec : CANCELED in 13m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/05520cae5aeb4084893cc175ea90f089 : CANCELED in 13m 16s\n\nWarning:\n  Comments left for invalid file octavia/api/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v1/driver.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v2/driver.py\n  Comments left for invalid file octavia/db/repositories.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/api/common/pagination.py\n  Comments left for invalid file octavia/api/v2/controllers/load_balancer.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/amphorae/drivers/health/heartbeat_udp.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/lifecycle_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/compute_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/lifecycle_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/compute_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/database_tasks.py\n  Comments left for invalid file octavia/network/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/common/policy.py\n  Comments left for invalid file octavia/common/data_models.py\n  Comments left for invalid file octavia/certificates/generator/local.py","accounts_in_message":[],"_revision_number":1},{"id":"581d7395be57157cd8c289b561befc367be74fc9","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-30 14:36:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c221d53620563856fd548e0f8c3c3f6630587aa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 16:50:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/17cc616069a14b19adedc0a833bc0f92 : SUCCESS in 10m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/233cf17dbb4a4adcb4c87fc264344849 : SUCCESS in 11m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8aad7d607df0465680dbe2049abb9c49 : SUCCESS in 13m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67088becef2e431a954a8532e5146d25 : FAILURE in 8m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd1fcd68a3ee4a00bf6ddef0d233807f : SUCCESS in 8m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3348879aa170450a9e849ea683489e65 : SUCCESS in 9m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/170082b5e950497ead2f8816ebb9de95 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3921f24c98c243228c50137e6db4023a : SUCCESS in 16m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5aed282b774453abc298ecdf1a42419 : SUCCESS in 7m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/06c24b2ef61346eeb49d2dd1442c634e : SUCCESS in 11m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7e584a2ffb934f949ba86bab218f53bd : SUCCESS in 14m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c4cc69179ede4018bb77119940c43db4 : SUCCESS in 8m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/23a1d35cab60415ab7735bd3e9087dda : SUCCESS in 11m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b04f354972ed4edd93ad222fb8711c09 : SUCCESS in 1h 54m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b6dbdebea2704f069143f18e8915b0fd : SUCCESS in 2h 12m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9892fd780b1b460dab8ee199421cfc07 : SUCCESS in 44m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/57fc9055fe2b4be68e404d53c4499a81 : SUCCESS in 48m 53s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e381c0c0dcc94a7db19b66b81312bff9 : SUCCESS in 1h 36m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a6e7d50621b3485ba75fed587a6fe85d : FAILURE in 18m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a61e8cfc650c490aac4badcf756bce7b : SUCCESS in 1h 45m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/06a89419d680466db26694d90b71439d : SUCCESS in 1h 28m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2f3608d67e214adcb9256f09517ebb30 : SUCCESS in 1h 32m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f329f4b6d30f4a1d9437833431bc6594 : FAILURE in 14m 06s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/api/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v1/driver.py\n  Comments left for invalid file octavia/db/repositories.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v2/driver.py\n  Comments left for invalid file octavia/api/v2/controllers/load_balancer.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/api/common/pagination.py\n  Comments left for invalid file octavia/amphorae/drivers/health/heartbeat_udp.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/compute_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/lifecycle_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/lifecycle_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/compute_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/database_tasks.py\n  Comments left for invalid file octavia/network/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/common/policy.py\n  Comments left for invalid file octavia/common/data_models.py\n  Comments left for invalid file octavia/certificates/generator/local.py","accounts_in_message":[],"_revision_number":1},{"id":"3705736a4baf0a22369742008b679b69a34742a8","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-07-01 07:23:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"954cf83c6414bc52e1ef477fe4bd8e79eb506429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 09:27:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ac279e85c68744d6ae091aa5941dae3a : SUCCESS in 6m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fca4988fb481422f9feff132d283468d : SUCCESS in 12m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92780dc7b8db49c89ef9b3554bb543e7 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c333dc0a2b0342c6821dad1112d2f11f : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88be511a04bb4e3a9b210e1bcb656c3c : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c093608d22c484da78c9e98e476dba1 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee2b8659143d48acb0026de99043910a : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f13e8a3af73445ed8c6ff5c7a9be1997 : SUCCESS in 12m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f38a0b95e18e461bae0bec32638ff946 : SUCCESS in 6m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f8b922665a76454b918a4e3a7ef52ccd : SUCCESS in 7m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b1ed674499ad4b8998e1725ac7961f87 : SUCCESS in 9m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3ac88ea4abc24172b3808df6d8d78d64 : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/25c903f2b5544aa78aecaae54c7d3e07 : SUCCESS in 9m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d37f19a25e1847208bafdb6c68c5e81e : SUCCESS in 1h 49m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c067f0a006084206ab03d6d28dd7e7c6 : SUCCESS in 2h 02m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f65550953cb843688e4ba1af1199cb87 : SUCCESS in 40m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd668aff68d148b1aefaa34f2abf46da : SUCCESS in 43m 30s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5117339338f7434dadfc35691f1cc99b : SUCCESS in 1h 59m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b8d40b4a30b9421487cebc8f989cf4ac : FAILURE in 19m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b6b4c5f0d0be4c9db49a2863adbd2f4d : SUCCESS in 2h 00m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d1904dd254b94f2f9310b3170e1285d3 : SUCCESS in 2h 01m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/397308211b8848db9b46cf51354f4f29 : SUCCESS in 2h 01m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ecc7af10f8d4487ca61e1935da3de63c : FAILURE in 11m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d2d976434ad7471e9954610fcaf96479563a1f9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 09:27:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1903f584e3a3082d2ea0a5897ea55fc7c8d38e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 11:28:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/33f63f9db2c54b038ea958e872fa11bc : SUCCESS in 6m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/102ea3fdd56e43c18466937467ad5d51 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/450efd826835417b99ec2ac750450f7c : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bf2d6a17274481cbf7f15a671ad4db7 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da59ab27e2df4a3cae7d5196839768eb : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f73dd5bb76244a70ae47d39d4d0c438a : SUCCESS in 15m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1db64e1fdf12403faf3fe819233d5a62 : SUCCESS in 7m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9a3235cc1b7c4a6c905d9bfb798ba34f : SUCCESS in 5m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/86083521dcb4453d8413c1e3239746e1 : SUCCESS in 9m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/caeb5c7935124b9abcb04ad4bcec0d9f : SUCCESS in 1h 50m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/991ca3d0d24141d38c6c3d9fdd1f9125 : SUCCESS in 1h 33m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/78cac8de636b4f59b07db16f4c07e859 : SUCCESS in 43m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d2e531f91824028a02d4fafe8fec536 : SUCCESS in 45m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b4d9e93ae1c54b36a59a505ea455308d : SUCCESS in 49m 03s","accounts_in_message":[],"_revision_number":1},{"id":"0f9747a7a8ef6ab95c2d36884e10d6bb05dc24b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 11:28:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"22fa827ad2ab66f9741c6cd2c9332697d6c4c434","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 11:30:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/e96548d0696f465881b48f9dff4ccc24 : SUCCESS in 55s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b61a6b11cc2e486482649b26ba1785a1 : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d9e3c1c9d4e4a3093d36c2acdb8f4b2 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b365ba489fe8254f32a29447c718cc9e948d491","revisions":{"1b365ba489fe8254f32a29447c718cc9e948d491":{"kind":"REWORK","_number":1,"created":"2021-06-29 06:00:43.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/16/798416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/798416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/798416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/798416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/798416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/798416/1"}}},"commit":{"parents":[{"commit":"d10c9c2f9afacff8047a8005843483a07fe7472e","subject":"Merge \"Fix using host_routes in VIP subnet with amphorav2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d10c9c2f9afacff8047a8005843483a07fe7472e"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-06-29 05:59:59.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-06-29 05:59:59.000000000","tz":480},"subject":"Changed minversion in tox to 3.18.0","message":"Changed minversion in tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 in order to\nreplace tox\u0027s whitelist_externals by allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nChange-Id: Ia815461a15bff065fd63e9ce80636e2b736bebdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b365ba489fe8254f32a29447c718cc9e948d491"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b365ba489fe8254f32a29447c718cc9e948d491"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
