)]}'
{"id":"openstack%2Foctavia~798577","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I5ed6288b2776bd7f1c9b67e9cfd9a8f05b1196be","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I5ed6288b2776bd7f1c9b67e9cfd9a8f05b1196be","subject":"Add amphora_id in store params for failover_amphora","status":"MERGED","created":"2021-06-29 08:54:41.000000000","updated":"2021-07-23 20:08:30.000000000","submitted":"2021-07-23 20:06:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798577","meta_rev_id":"d14d1ad16a5400a76a54b3ab61c86bdba897af38","_number":798577,"virtual_id_number":798577,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-23 20:06:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-06-29 09:34:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-07-23 18:25:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-07-23 18:25:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-29 08:54:41.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-06-29 11:14:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-23 18:25:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"a2d26b8f6ae7856c43c94fc8f45814d7181ade60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-29 08:54:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2955b2b9d31eee97f7ae7b6ad0f4e8ddc7775d8","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-29 09:34:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"deb9de33ee290b8010c44174bbdf42bfbfe29d3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 11:14:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c770d7f8cf574c1cbc80fae1d5d128f2 : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/255124d12ad0433e960ed256e66464a5 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb06f2160083456c8f4e7e3ecf25eeee : SUCCESS in 7m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db84d33b850f4d7b98fa6036f0b374aa : SUCCESS in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ba1babc8ca84b14ba9442d22f13dd4c : SUCCESS in 14m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/acd9d7611e504e3ea170ca497ca25860 : SUCCESS in 8m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2d2a4e2dc07342348a8c160a2c5c711b : SUCCESS in 9m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/881c08c41d854098af9788f7b0e41186 : SUCCESS in 9m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2419c443bc054c5ba35a3c3d60714e4f : SUCCESS in 1h 24m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ec6cd4ff04741b596e3dc0ac06379b3 : SUCCESS in 1h 37m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/156980138e3d46c9b7cd0c81b4ae789d : SUCCESS in 44m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/028d60654abf462c8cf49784c483ca9e : SUCCESS in 59m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/137ca55e8d8e4e3891f4fdbfb60e06b1 : SUCCESS in 48m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/35c087fd640a4a4ea856ed93ae46721b : SUCCESS in 1h 01m 47s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b4e34f3e4a434138ab5f0ff0444a5a6c : FAILURE in 2h 07m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/483800b9067743cab29595016144988d : RETRY_LIMIT in 8m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f18385d019b7462188c6b6be8fe5bbff : SUCCESS in 1h 55m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"80a49d415c3755c33788f79e6bc22a132b004e39","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-23 18:25:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"4f67ea1ab09b2f3c9c70c53ba01e890005c69fa8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 18:27:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"990babe4e1b8857b705a1a544375570bb9365b8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 20:06:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac73063e32c3480e8d3612c277343fcc : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3075a4177154440699800b2e7f6dee04 : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9eff37e0aeb64dd0a507e84ace859fbe : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1ad396381464f3c92c7dd016ca45cd7 : SUCCESS in 18m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/99fed71ed6c74a3d9570fea595555a47 : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d8796ac9e6d142bfbeef7876787a62ed : SUCCESS in 1h 27m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31e165fd23a2495398aee2c96c3df15c : SUCCESS in 1h 24m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1fffb8965fdc497fbcd1faaabe65b4ad : SUCCESS in 43m 20s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c88dd845fa9a433783b0937454015910 : SUCCESS in 41m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ddefab368334e1dba863a4e3c8b3ed3 : SUCCESS in 47m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/712f60be186e4b32ade7741f3b102518 : SUCCESS in 1h 21m 18s","accounts_in_message":[],"_revision_number":1},{"id":"76797fbf8757fbcd96a4c2fc371b97e56ad14cc2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 20:06:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d14d1ad16a5400a76a54b3ab61c86bdba897af38","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 20:08:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/119dd06c499644e1ae5280840f98676c : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de60278205a0531906a44cbebb33576ae24c23df","revisions":{"de60278205a0531906a44cbebb33576ae24c23df":{"kind":"REWORK","_number":1,"created":"2021-06-29 08:54:41.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/77/798577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/798577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/798577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/798577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/798577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/798577/1"}}},"commit":{"parents":[{"commit":"c0af684fd701706a9fcf80a64991f070132591b7","subject":"Merge \"Fix empty Batch Member Update to unlock objects\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c0af684fd701706a9fcf80a64991f070132591b7"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-10-29 14:46:34.000000000","tz":240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-29 08:54:25.000000000","tz":120},"subject":"Add amphora_id in store params for failover_amphora","message":"Add amphora_id in store params for failover_amphora\n\nSeveral tasks require amphora_id parameter to be passed in\nget_amphora_for_lb_failover_subflow.\nExecution passed results in error:\n\ntaskflow.exceptions.NotFound: Mapped argument \u0027amphora_id\u0027 \u003c\u003d\n\u0027amphora_id\u0027 was not produced by any accessible provider\n(1 possible providers were scanned).\n\nAlso fix getting ID parameter from amphora dict in\nAmphoraIndexListenersReload and add missing retries for create_*\nresources in v2 worker.\n\nChange-Id: I5ed6288b2776bd7f1c9b67e9cfd9a8f05b1196be\n(cherry picked from commit 1a154839c2dfb5a9c89963b27a96f0adef1038ee)\n(cherry picked from commit ee7b262199840bca7f08a111ea535392396bb416)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de60278205a0531906a44cbebb33576ae24c23df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de60278205a0531906a44cbebb33576ae24c23df"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
