)]}'
{"id":"openstack%2Foctavia~798811","triplet_id":"openstack%2Foctavia~master~I4651f8bdaca13486a1dc6408f445d41779a95653","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I4651f8bdaca13486a1dc6408f445d41779a95653","subject":"Fix new pylint issues","status":"MERGED","created":"2021-06-30 09:16:54.000000000","updated":"2021-06-30 18:18:24.000000000","submitted":"2021-06-30 18:17:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"798811","meta_rev_id":"db347faecfc0cd2732a254cbfc888abd9a46ff6c","_number":798811,"virtual_id_number":798811,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-30 18:17:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-06-30 09:29:50.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2021-06-30 14:02:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-06-30 14:02:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-2,"date":"2021-06-30 09:29:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-30 09:29:50.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-06-30 11:40:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-30 14:02:58.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-06-30 16:29:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"}],"messages":[{"id":"895480d0f837783c89c8ab8805d5d883ce89b3f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-30 09:16:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91d7af1b3bb13e04ed6eec8fb461c2d393532bba","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-06-30 09:29:50.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2\n\nFix in stable branches should be by setting an upper-constraint version to pylint, hence BC-2.","accounts_in_message":[],"_revision_number":1},{"id":"350d75b804185271f6ac2ea3ae2a972074138232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 11:40:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/012e39ea3c174c3fa8e3e2b261b9b2e9 : SUCCESS in 12m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a36f018d2b9e49649c452fabb336330f : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68e3131d40834124a7b8fe6d04cce5b4 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc1b1a7da32245269230784365160c5a : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c788f29d32d84d66850c49b872342a9a : SUCCESS in 8m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4b9cf8ebe824276b69c421c69fa530a : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c008752df4044687bef67dee3231aba7 : SUCCESS in 13m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8e9f526f4b1d4a42931d09b7517c7d5b : SUCCESS in 8m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d7abd5b0ef7b49c69e3325313fb46603 : FAILURE in 13m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/29ecafe54cc6447a8049d0a3f73ffd75 : SUCCESS in 7m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c9eba186f6c64ca0b41e4f7bdb9789fc : SUCCESS in 9m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4e66edfc2e4646998e296952b10b11cd : TIMED_OUT in 2h 15m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1939645b5af44000a365efe2716fddb0 : SUCCESS in 1h 41m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/64ee49f717344f3ab5d70a395d0a4bbc : SUCCESS in 43m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/00b1af061d2a492f920f886aab3ff257 : SUCCESS in 1h 05m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bec3d1f0f9904adb81fe4eea46ab6da0 : SUCCESS in 1h 08m 32s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a74804bd3a1e42d08fb6a134b975ad2b : SUCCESS in 1h 57m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1a88f49de3af4d7c801ed668f37554bc : FAILURE in 19m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ca42cba162264ffc8b1e0e5124d416f5 : SUCCESS in 1h 53m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/74b93f4d263c42d79240d13887f587de : SUCCESS in 2h 22m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/29c030a6752a4b639ec3ec391ebf1f0c : SUCCESS in 2h 01m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f67a480eec3e4b4f8009285b335842c6 : FAILURE in 15m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7c9129abc32066ffcada2f0480fc324378b8aa8b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-06-30 14:02:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"ffe1763f5d87080f3385758269c44d1e1028484f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 16:16:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54ce98bf105547f98598a6af36cb3d95 : SUCCESS in 13m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc3cb1a395a145a1aed4d22187d2e1dd : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/435cbdc0d6634feb849e35cbb19ab606 : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61a05eebe42f4447a3baddcdc8913de1 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/402f433bc0e8424eb308a4fd6cde8e52 : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8024cf427c7543dda84bd938c08b0451 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85110600dacb46d6b4dd1dd931fceada : SUCCESS in 14m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fe87088398004a48b9043fe9ff973892 : SUCCESS in 7m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/645bc960cde94440b975bf5e74418d3e : SUCCESS in 10m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3a8740d96af14efcb19d08720a5776de : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5345204e24a84a919b8a48ceeb999fb5 : SUCCESS in 14m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/787c59d933514dd1b37cd95493088c53 : SUCCESS in 1h 52m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/03897a5292ac40ceb1b23c21914193c5 : SUCCESS in 2h 09m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b621991e00f34b12ae9ca32183219d84 : SUCCESS in 49m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec8e71d1bd6448d884a60b34f765931d : SUCCESS in 46m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7a024e021ede45418c86cc7ed388987c : SUCCESS in 51m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5e55de0ebe45488ea90e5ce56e655d2b : SUCCESS in 1h 56m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/72626981190c4410b07da03f968535b8 : FAILURE in 20m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ca9d80b2058e43cb84371858caf8e2af : SUCCESS in 2h 07m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ebfa890d6b0d40f793baa7d649e0ed55 : SUCCESS in 2h 12m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ee90ffc7c04645c396a6c599f161d09c : SUCCESS in 2h 05m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e15abdee623b47d1afa5a52ddf315f7f : FAILURE in 14m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f570ebb19c0e5671ab9dc49683dee1af6d7e4ffa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 16:17:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af7d599af415f8eb6af3bf64f4c67fb86a211d40","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-30 16:29:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2b3ca43dc3ccca549651233800682ef9cd3083a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 18:17:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/375014e978d7499da0a6cbdeb739c7f0 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/340bebe8b1994269af0dd52662bc5710 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cdddcab39b6427faf232cba56c6fc5e : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99cdbfb6b28c43a5af261020f29e5d74 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c67f79a57f9422d82d9fc4a392675a7 : SUCCESS in 11m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6ff7f243b1774951a06cb946993258a5 : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a1572b1f9061489a9baff1ed2e8dc5e7 : SUCCESS in 8m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ec088e88e39a489e8b1179eb89585cd4 : SUCCESS in 1h 46m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2153c92d780a4d08a87bdf81724c05fa : SUCCESS in 1h 58m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cd856f70e9374390b3aa8645644a30d1 : SUCCESS in 47m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da9fb77ca46f4738b96a99fa80006748 : SUCCESS in 44m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/359aba07500f4f2ba7a4b2c4169b250e : SUCCESS in 1h 07m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f28e3f500cf4ac31e2827ebce3d8a17564b04d2c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 18:17:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"db347faecfc0cd2732a254cbfc888abd9a46ff6c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 18:18:24.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ebe147499a44510be7f307abd852291 : FAILURE in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a9de5b27a33a5521d0e0a1fdc4c63eb439d8149","revisions":{"4a9de5b27a33a5521d0e0a1fdc4c63eb439d8149":{"kind":"REWORK","_number":1,"created":"2021-06-30 09:16:54.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/11/798811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/798811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/798811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/798811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/798811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/798811/1"}}},"commit":{"parents":[{"commit":"d10c9c2f9afacff8047a8005843483a07fe7472e","subject":"Merge \"Fix using host_routes in VIP subnet with amphorav2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d10c9c2f9afacff8047a8005843483a07fe7472e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-30 09:06:56.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-30 09:07:03.000000000","tz":120},"subject":"Fix new pylint issues","message":"Fix new pylint issues\n\nnew pylint 2.9.0 provides some new checkers and detects new errors with\nexisting checkers:\n- consider-using-enumerate (disabled)\n- consider-using-from-import\n- arguments-renamed (disabled)\n- consider-using-with\n- consider-using-dict-items\n\nChange-Id: I4651f8bdaca13486a1dc6408f445d41779a95653\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a9de5b27a33a5521d0e0a1fdc4c63eb439d8149"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a9de5b27a33a5521d0e0a1fdc4c63eb439d8149"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
