)]}'
{"id":"openstack%2Foctavia~799405","triplet_id":"openstack%2Foctavia~master~Ifa8c620fe76517b1c040af5905e6a6c1cdb4c922","project":"openstack/octavia","branch":"master","topic":"fix-oslo-policy-warnings","hashtags":[],"change_id":"Ifa8c620fe76517b1c040af5905e6a6c1cdb4c922","subject":"Fix oslo policy DeprecatedRule warnings","status":"MERGED","created":"2021-07-03 23:59:46.000000000","updated":"2021-07-15 09:11:16.000000000","submitted":"2021-07-15 09:09:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"799405-fix-oslo-policy-warnings","meta_rev_id":"77d48f48377f2735b344a117b0aee4a089309ffd","_number":799405,"virtual_id_number":799405,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-15 09:09:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2021-07-12 21:23:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-07-12 18:13:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2021-07-15 07:05:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2021-07-15 07:05:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":-2,"date":"2021-07-12 18:13:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-04 02:33:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-05 05:43:56.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2021-07-06 16:55:25.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-07-06 16:55:55.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-07-06 19:08:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-07-08 05:39:34.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-07-14 20:29:36.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"db13bcfb5f32159deb7b9a56e01d80e727d51496","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-03 23:59:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c48ae4691642124ce11c11b1232ba9e0412773d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-04 02:33:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4cdd2e4d005425e8b8eb58a70f9a4bb : SUCCESS in 12m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5246e8144d84777bcbd0cd435e52f2c : FAILURE in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51db1335f37a4112b96252b2aa73831f : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecbf0ccb197e4531805ee4c81d0cdbf4 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/558ece65e30a4837b77582fd351deffb : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9472907a2eb349ee8c71e1c8890e0937 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8c13825cf774091b71ae98facf9d8a0 : SUCCESS in 15m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/79a1a1c758044595947332bfdeb20bd8 : SUCCESS in 5m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8f18136237e94afe91c5b55d43177fd1 : SUCCESS in 8m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bc14dc96518a4a338cf217e39d01ce34 : SUCCESS in 5m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7951ff9d6ce647fd91b49d17de49a20f : SUCCESS in 8m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3563364b70044552ad08025ff9eac91c : SUCCESS in 1h 26m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e3bb860eaa34979bf61d8a7d8caf8db : SUCCESS in 1h 35m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/190836e5e3ac4ac68815cbce5e130fec : SUCCESS in 36m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5804e0eae67c4b6c903c4c79f39e2d72 : SUCCESS in 38m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9e9bb33895d8479fb64103810c206327 : SUCCESS in 1h 23m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/10e3256f835541af99315833fa585a3c : SUCCESS in 1h 55m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c248377f2fdd4f7fadd30e8eb5ba4b3a : FAILURE in 13m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8c40377485c446ef81fe2a9369a4af38 : SUCCESS in 1h 57m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c606a3cfbcd447c28fddd25ceb7608c2 : SUCCESS in 2h 28m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/be9e78a4b1e34e32b9296f97122f5d1a : SUCCESS in 2h 02m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/70c556229c9046a4a56e3f1a5e04b486 : FAILURE in 9m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb17bd59f296697d79b59b3a3019034ad9f19b11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-04 23:02:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0456e2421312001c2f8780f5a05525636f2260be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 01:11:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1c9ae29184c1483c93b5add6e5be6803 : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf4aa9ecde2e44ff94d7b2841143fdd5 : SUCCESS in 11m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00503fc1f835467cb56e00a9966c279c : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71edb515769c4f55a21e8dc9af7d2f36 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e05257b0ae04eedbbf5698149e09c77 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d540dddbdcd459f86fe0c9a8ebf1067 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/398125b76c224533a84de9988a60be0f : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/010decb007cc42c8b6544687fa0813d2 : SUCCESS in 13m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/47890a0415f34f39836ca331629d3cf7 : SUCCESS in 6m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b89d3fa1dc884066b52b60d83aad8a51 : SUCCESS in 10m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/528ff5dde853499a862e834bc2ac3354 : SUCCESS in 5m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/af80fbbf68444e9c9dfbcda9261e6974 : SUCCESS in 9m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7f3c638211c644c385c3a002409d0fdb : SUCCESS in 1h 48m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cab9b51a3a9c4096bda7169c7807e6d2 : SUCCESS in 1h 27m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7b82cbaf518f4ce68561ef8aa2d3ee56 : SUCCESS in 40m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/547cf9698223476dbca9e6f6b644279e : SUCCESS in 39m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a6814f35f4324756b073dd39e4fd45ee : SUCCESS in 1h 02m 43s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/22593e9ac8d84727a6f8630cbaee3d91 : SUCCESS in 1h 58m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b8c540e8f9b44591842ef82d4101243c : FAILURE in 13m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b19eeeaf170e4e9ba0cce92088ca962d : SUCCESS in 2h 00m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/585bc5e1b90e4a6bb07adcfe56525e69 : SUCCESS in 2h 06m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/42bc7a085c1c46bb99731c0bb8720869 : SUCCESS in 1h 49m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a83d258f075d4e5f892db255d2592d0f : FAILURE in 7m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f2b408c7d8d93995307a58858b94ebe604a7ca8f","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-07-05 05:43:56.000000000","message":"Patch Set 2: Backport-Candidate+1","accounts_in_message":[],"_revision_number":2},{"id":"892a203e6a29c3c4f2471b5604bacb0ddecd23f7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-06 19:08:03.000000000","message":"Patch Set 2: Code-Review+1\n\nNot sure about backporting this as it\u0027s bumping requirements, which we don\u0027t typically do in stable releases.","accounts_in_message":[],"_revision_number":2},{"id":"e3733ca678fc8ce7c19f58a93516367e8c4f26c2","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-07-08 05:39:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"aee13e45318dab7e34b022084244ee23e3ee17bd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-12 18:00:16.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nJust one question, I think we missed one cleanup.","accounts_in_message":[],"_revision_number":2},{"id":"fdc640f0b8cba4f51a6c084b313b0916b65725da","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-12 18:01:47.000000000","message":"Patch Set 2: Backport-Candidate-2","accounts_in_message":[],"_revision_number":2},{"id":"5255cdab4feb7a00a5d881c918c352ce4a9a294f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-12 18:02:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4f0af9521c774bfda65fc51f4de1364155735dfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-12 18:10:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"344629e23e4beed4e3dcbd7bc9c2a9712d9c3718","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-12 18:13:50.000000000","message":"Patch Set 3: Backport-Candidate-2 Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"7d775c882ac90cbc1158d43bab4a179a3440ef4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 20:50:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d31500d8b6ff4ff9846277acafd5a588 : SUCCESS in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bb645d27bf24c86a5d9e499ace5e304 : SUCCESS in 10m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/353f3100fe5141a194cd080071a05949 : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d46b92f0feaf404ba081c08210173e6b : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70706dfc918641348d7f58d88dcbeebb : SUCCESS in 11m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d4b4500e6664bcab5176a7194a21323 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d569c7518202416495f6fb5ce161077a : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6d2419cf6e84052ab617fe8001e2142 : SUCCESS in 12m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2695823bce9d4a5386bb55ec4d09c750 : SUCCESS in 6m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e6f4b1b886694d8b94a63d47a78cb5d0 : SUCCESS in 9m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/443f5adda30047ef89f34e9f0bb09413 : SUCCESS in 7m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3c1568a61abb4e0a9946280daa697137 : SUCCESS in 7m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a35ad12ec10140edb911c290c130569f : SUCCESS in 1h 40m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6a40cdae31e94b6fab4013dc38089ea3 : SUCCESS in 1h 36m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1355848dc963462396178f0849363a7c : SUCCESS in 37m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1d9e8992ec884dd7b2bf50c4ed5250b9 : SUCCESS in 37m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0923d72bc4ec438d8554429945088665 : SUCCESS in 1h 03m 14s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f20182543a6642ba96eec3d26730fcfe : TIMED_OUT in 2h 35m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2a3cf54674a94d8799b21cea389da04c : FAILURE in 14m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3310fd8f41784368937fbba19475b66c : SUCCESS in 2h 02m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2eea80e3b00d4146915a01dd710a8e5b : SUCCESS in 2h 32m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/26fa22bd28e54ef6a9271c383f273ba3 : SUCCESS in 2h 01m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e351aecaa5d24cc3a52b1d0c772aed81 : FAILURE in 16m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"69753c26ca5ce986274344f14729b8bc0a166528","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-12 21:23:18.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"771429b827258f885ad659c1f6e27c3af59dc51a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-07-15 07:05:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"be7414b69b7df4651b56c68425628c7165d96753","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 07:06:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8da2d6798b6f0d13953452a66d0cbcab5b674036","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 09:09:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/88f47218176d43ee944e7f5970f63002 : SUCCESS in 3m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54071badb1694fd2aa3cfbba2f4bfab6 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28d541981b8f4b168c0f3dfe576f7d53 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e7afb33e3534b049f580f7f21dd4ca2 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d74161232a34446b9f0c532af48729ea : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd7d3a1c91984aee917766c3f857a91e : SUCCESS in 11m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8f41a7ef104c49068011b51d8b18bc7c : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d4c2662578d348d896174a029aa916d8 : SUCCESS in 7m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d38883fa31d2465caf17214629d644ec : SUCCESS in 1h 42m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/58e2a7cffa4e4b61b4707471fdb5d8b1 : SUCCESS in 2h 01m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7b6f9a5a7ff94939aa04395895ef77d4 : SUCCESS in 53m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b53809f25ef9498fa677d508ba4856f8 : SUCCESS in 51m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/74ff7a652ab940b3928ea7b5729a3d82 : SUCCESS in 58m 59s","accounts_in_message":[],"_revision_number":3},{"id":"97f663dd04701dd910d6999ad25626db3eb22bde","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 09:09:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"77d48f48377f2735b344a117b0aee4a089309ffd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 09:11:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d8d42918a0c460191a9dd68cde004d8 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7e264e1f21ee396d046e0df0c1a4f4218039e4aa","revisions":{"625a025222878b05733057d5efdda90503fbf168":{"kind":"REWORK","_number":1,"created":"2021-07-03 23:59:46.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/05/799405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/799405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/799405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/799405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/799405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/799405/1"}}},"commit":{"parents":[{"commit":"e647f6d71a8479b2ce5fd8c2efb27b97c1f54505","subject":"Merge \"Edits for \"Basic Load Balancing Cookbook\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e647f6d71a8479b2ce5fd8c2efb27b97c1f54505"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-07-03 23:59:30.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-07-03 23:59:30.000000000","tz":-300},"subject":"Fix oslo policy DeprecatedRule warnings","message":"Fix oslo policy DeprecatedRule warnings\n\nSince 3.7.0, oslo policy started the DeprecationWarning[1] if\ndeprecated_reason and deprecated_since param are not passed\nin DeprecatedRule or they are passed in RuleDefault object.\n\nThese warnings are logged for every test which increase the\nlog size and sometime can full the log buffer and fail the\njob.\n\n[1] https://github.com/openstack/oslo.policy/blob/3.7.0/oslo_policy/policy.py#L1538\n\nChange-Id: Ifa8c620fe76517b1c040af5905e6a6c1cdb4c922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/625a025222878b05733057d5efdda90503fbf168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/625a025222878b05733057d5efdda90503fbf168"}]},"branch":"refs/heads/master"},"6bc985fbbdc55de96e994bdfb7f79b022f55d281":{"kind":"REWORK","_number":2,"created":"2021-07-04 23:02:16.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/05/799405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/799405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/799405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/799405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/799405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/799405/2"}}},"commit":{"parents":[{"commit":"e647f6d71a8479b2ce5fd8c2efb27b97c1f54505","subject":"Merge \"Edits for \"Basic Load Balancing Cookbook\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e647f6d71a8479b2ce5fd8c2efb27b97c1f54505"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-07-03 23:59:30.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-07-04 23:02:04.000000000","tz":-300},"subject":"Fix oslo policy DeprecatedRule warnings","message":"Fix oslo policy DeprecatedRule warnings\n\nSince 3.7.0, oslo policy started the DeprecationWarning[1] if\ndeprecated_reason and deprecated_since param are not passed\nin DeprecatedRule or they are passed in RuleDefault object.\n\nThese warnings are logged for every test which increase the\nlog size and sometime can full the log buffer and fail the\njob.\n\n[1] https://github.com/openstack/oslo.policy/blob/3.7.0/oslo_policy/policy.py#L1538\n\nChange-Id: Ifa8c620fe76517b1c040af5905e6a6c1cdb4c922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6bc985fbbdc55de96e994bdfb7f79b022f55d281"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6bc985fbbdc55de96e994bdfb7f79b022f55d281"}]},"branch":"refs/heads/master"},"7e264e1f21ee396d046e0df0c1a4f4218039e4aa":{"kind":"REWORK","_number":3,"created":"2021-07-12 18:10:39.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/05/799405/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/799405/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/799405/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/799405/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/799405/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/799405/3"}}},"commit":{"parents":[{"commit":"e647f6d71a8479b2ce5fd8c2efb27b97c1f54505","subject":"Merge \"Edits for \"Basic Load Balancing Cookbook\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e647f6d71a8479b2ce5fd8c2efb27b97c1f54505"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-07-03 23:59:30.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-07-12 18:10:27.000000000","tz":-300},"subject":"Fix oslo policy DeprecatedRule warnings","message":"Fix oslo policy DeprecatedRule warnings\n\nSince 3.7.0, oslo policy started the DeprecationWarning[1] if\ndeprecated_reason and deprecated_since param are not passed\nin DeprecatedRule or they are passed in RuleDefault object.\n\nThese warnings are logged for every test which increase the\nlog size and sometime can full the log buffer and fail the\njob.\n\n[1] https://github.com/openstack/oslo.policy/blob/3.7.0/oslo_policy/policy.py#L1538\n\nChange-Id: Ifa8c620fe76517b1c040af5905e6a6c1cdb4c922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e264e1f21ee396d046e0df0c1a4f4218039e4aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e264e1f21ee396d046e0df0c1a4f4218039e4aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
