)]}'
{"id":"openstack%2Foctavia~803426","triplet_id":"openstack%2Foctavia~master~I8fd50b3dac8b3c211101e9257a18b3376f314a91","project":"openstack/octavia","branch":"master","topic":"dnm","hashtags":[],"change_id":"I8fd50b3dac8b3c211101e9257a18b3376f314a91","subject":"DNM: Checking gate","status":"ABANDONED","created":"2021-08-04 08:11:46.000000000","updated":"2021-08-18 14:17:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0aef697514e26589e1d2009b1c39195ce92d9cda","_number":803426,"virtual_id_number":803426,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-04 10:18:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-04 10:18:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3685812767948167688016d73fbebf5f953c9144","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-04 08:11:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9f824e69efc408e993051d22064d71cdcf6559a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 10:18:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bf4a648c2f244ff9b09cfc5c90d1b50 : SUCCESS in 11m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55d0d3f8f6014d748b8a80e2e4a7f9cc : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cfbb45d1ad04328ae04d508a12c8dc2 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9025a3fd3a0745fc899a4a1b644f4738 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02990ba80393404a84544cee577fa703 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18de8565e28c4c2585437b90f071f7a6 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7e0bfb465a24d80b7bb7ae05b7df0e9 : SUCCESS in 11m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b60e46905a4340d6978307453bba8e8d : SUCCESS in 6m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/93f25d56006d47ae921792868064fa6b : SUCCESS in 9m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/64b1ea2fc76c4d428cc11abd91062f60 : FAILURE in 5m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d35d15e7537f42a084311a64deb568e6 : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/20ecc3fbfe064303a3424fa77b94edac : SUCCESS in 1h 40m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d4ebec60863c456bb880897bc635fe08 : SUCCESS in 1h 44m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0a8ccaa8f11a468f81351a291e2a7859 : SUCCESS in 39m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba22ffff019e4caa8a625e52cee84fbb : SUCCESS in 42m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9966e9b61e5c4709a8533018cbfa4fc5 : SUCCESS in 1h 21m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7a54b8912c4e4ca694d7bb60498893de : SUCCESS in 1h 42m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0a2aff4eacc7423480f930bfd30372b4 : FAILURE in 18m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8029bc40a06f4c53b91316f602621956 : SUCCESS in 2h 04m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4a402c23f8654cb7be0bc19ecf05b95e : SUCCESS in 1h 22m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e2e6f4004d8145b69e73550046d215c4 : SUCCESS in 1h 25m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fa32e0ba70514f77b5658016755dff0d : FAILURE in 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0aef697514e26589e1d2009b1c39195ce92d9cda","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-18 14:17:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c4a7eaf643308f70a437e86675806ac389915d3","revisions":{"5c4a7eaf643308f70a437e86675806ac389915d3":{"kind":"REWORK","_number":1,"created":"2021-08-04 08:11:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/803426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/803426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/803426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/803426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/803426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/803426/1"}}},"commit":{"parents":[{"commit":"7e264e1f21ee396d046e0df0c1a4f4218039e4aa","subject":"Fix oslo policy DeprecatedRule warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e264e1f21ee396d046e0df0c1a4f4218039e4aa"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-04 08:11:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-04 08:11:32.000000000","tz":540},"subject":"DNM: Checking gate","message":"DNM: Checking gate\n\nChange-Id: I8fd50b3dac8b3c211101e9257a18b3376f314a91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c4a7eaf643308f70a437e86675806ac389915d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c4a7eaf643308f70a437e86675806ac389915d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
