)]}'
{"id":"openstack%2Foctavia~803461","triplet_id":"openstack%2Foctavia~master~I6ae2740557f0e62b6805e5dadd025e7bf44ec57d","project":"openstack/octavia","branch":"master","topic":"pip-debug","hashtags":[],"change_id":"I6ae2740557f0e62b6805e5dadd025e7bf44ec57d","subject":"Add caps for pip-extra-reqs/pip-missing-reqs","status":"MERGED","created":"2021-08-04 11:00:13.000000000","updated":"2021-08-11 09:49:08.000000000","submitted":"2021-08-11 09:47:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"803461-pip-debug","meta_rev_id":"e9e2cc8a3378cc28edc91af388a484f5973ed081","_number":803461,"virtual_id_number":803461,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-11 09:47:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-08-11 07:59:22.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-06 14:46:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2021-08-11 07:59:22.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2021-08-11 07:49:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-04 12:59:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-04 17:30:44.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2021-08-06 08:45:58.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-08-06 14:46:05.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-08-11 07:59:22.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"20325e834c62c516bdefbeddcb1e717308b382b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-04 11:00:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"490605ff2d6ef64d653e5275e1ac5783073be6fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 12:59:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a0c56ecd936e4dc5abeed0df36e200b7 : SUCCESS in 6m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c604d26c49f140edb893900d72092d7d : SUCCESS in 11m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/647e6d2cd7674bf7b08719d885d2c7c1 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/499e35f497e64e4b8616ac0f04b8fa07 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9abde5a020d244a295d0990bb5fc67f5 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af5cf8f86d0842afa89e6499714827ca : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46122b9fb9ea4314ba643ad41dac4944 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8229ed499fd14f9ba63b2f61136f1673 : SUCCESS in 11m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01ec11ba679d4e8686862680c5dd955e : SUCCESS in 5m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d562cd049327464e890ddd6ecdb0990d : SUCCESS in 7m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ccdc6cc843914295982807e99dfe5a97 : SUCCESS in 10m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5c6af9eb8bcf4cc790f862e47a0a186f : FAILURE in 5m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6900c4c4d8b54860b584fcde01d913b9 : SUCCESS in 9m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/90930ab94c6446138750bdba0c73df5e : SUCCESS in 1h 45m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd7684c7700543fca6182cd3bcf53a28 : SUCCESS in 1h 13m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bae69473c46a40838e48d9683339840e : SUCCESS in 34m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/63264f34b1554d7496c4b8acaffb4ee8 : SUCCESS in 31m 13s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b3cafd4506494ea9b9291d530d9172bc : SUCCESS in 1h 17m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3025168112674aef97824ac1821c3d36 : FAILURE in 14m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/82b929c941e64554aa1261fee3f9e3b6 : SUCCESS in 1h 25m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f417adb8d1ce4627990539e76010567d : SUCCESS in 1h 33m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/542ae61c09f742db8a1a01a00f3b153e : SUCCESS in 1h 15m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7383d8d57bfa4c97a2a8d056f64da468 : FAILURE in 10m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c872973dc7044b3a32db6a6bcdcbd93affcba65d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-08-04 17:30:44.000000000","message":"Patch Set 1:\n\nit looks like this commit in pip-check-reqs\n\nhttps://github.com/r1chardj0n3s/pip-check-reqs/commit/e48587750152f4edb8e5ff97429e9f6a98489973\n\nbreaks this job on ubuntu focal","accounts_in_message":[],"_revision_number":1},{"id":"929a0b1b3cd8c51ea25e52f727a5b7675b69aa83","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-08-05 05:35:48.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e it looks like this commit in pip-check-reqs\n\u003e \n\u003e https://github.com/r1chardj0n3s/pip-check-reqs/commit/e48587750152f4edb8e5ff97429e9f6a98489973\n\u003e \n\u003e breaks this job on ubuntu focal\n\nFYI I opened an issue in pip-check-reqs: https://github.com/r1chardj0n3s/pip-check-reqs/issues/66","accounts_in_message":[],"_revision_number":1},{"id":"75af24d887de3b531cfc44411903c1d6c281c1c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-08-06 08:45:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fbb4afd002a5ce4e07a97cb24e2f07d95ee4b886","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 08:55:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/912551c96438427dbb2c2a12ec172f06 : SUCCESS in 7m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/14d6178be63d4e2fa75732a3c8ae68a2 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":2},{"id":"fc2368f465eb15faaccecd8b1adf3b90e4bf6dfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-08-06 10:05:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"950506380c341ea4c11ad4c0415894fa9e405f30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 12:12:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/21f353fadd904a88a5f5dc20beaf3351 : SUCCESS in 6m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c2e2f16ff304c1d8b6107c1751c3d23 : SUCCESS in 11m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c66288cc9ca4100846359cfef5c9197 : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/260045bc99d14486b8d757a49a552506 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb9dc9b1799b4cbd967f2d95ead50e01 : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/420490cb3150450596520ad8a312303f : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50b48187afdd4f2da8dd41fb644fceda : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f6e8a4e8574de58a092b180166a55a : SUCCESS in 12m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/764e17619a214108aecefeed4f197e26 : SUCCESS in 5m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dac5d78f621041ab90b2ecbd83c37d67 : SUCCESS in 6m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9cea5d846ee540e1928ea7bed0591c94 : SUCCESS in 9m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d10ebb66a09f477794c7e7e5f3189acd : SUCCESS in 5m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7d201be4d36345bc880a83d596472e5b : SUCCESS in 7m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e2f8bc41a57d40c19e1380c749d5fda6 : SUCCESS in 1h 41m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/149c823de1c04d7bb5fff62144cc995b : SUCCESS in 1h 29m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/80aca49737c64187b151b865f6a39b9a : SUCCESS in 43m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5e40a71b335041a49220d4a4133a222f : SUCCESS in 34m 02s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4f8350c8f3de4803b6fec3b47f3bb14a : SUCCESS in 1h 44m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/58f94ebd5d7c429d94d7b37b3e2bdd1c : FAILURE in 14m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7e726838d1444e57a5692723d14137e5 : SUCCESS in 2h 03m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bf07daca7c964778a9b7be6945629ee3 : SUCCESS in 1h 47m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ec106fb78c8641faba3f6b515c80392a : SUCCESS in 1h 30m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c326ee6487e84b698ece8951ec3556b9 : FAILURE in 8m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d780c895fe5bfcff82301f228f4c24a72416000f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-06 14:46:05.000000000","message":"Patch Set 3: Code-Review+2\n\nI am ok with this approach.","accounts_in_message":[],"_revision_number":3},{"id":"cd45957c552184d2251ad112ed06291759654d22","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-08-11 07:49:49.000000000","message":"Patch Set 3: Backport-Candidate+1\n\nBC+1 because it also affects stable branches","accounts_in_message":[],"_revision_number":3},{"id":"e5e009c09165572725581dcd499a83ac8da765b9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-08-11 07:59:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8e34fb38faf21ae06b9c56d5aa402ea682ddb21c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 07:59:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4efd1c7525b5f5562bee443a15c4912b976a4751","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 09:47:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c31e3a029d7e44a6bc9675f3ffd3e810 : SUCCESS in 7m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97d6f4c5f4f9457385ce1bc8ddc80860 : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98b881dfc1c2479d944ed6d42243a860 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb913553e33d4e84a1682342bf49bf31 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab649ad8ddec4d569585126546f1ab39 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30295e23dbb6431b88acdedde48f3325 : SUCCESS in 13m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e58aedee63541e08543b917ca29ec7b : SUCCESS in 6m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c0f4eb530f5b4d418d0ab00d1def6717 : SUCCESS in 5m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/88126d51275d44b9b184d117c053fe63 : SUCCESS in 8m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9c5a2c99d5a94290975f419b87c066c3 : SUCCESS in 1h 43m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/010e30590a0d41e3b521ca884b8faa7e : SUCCESS in 1h 24m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a7f82f237caa456481566fb634ae2da0 : SUCCESS in 35m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d70d56483d2a495cb3e4b5ab81ce79a0 : SUCCESS in 34m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/56df355dbea5408fab63c8a0b1d767f8 : SUCCESS in 48m 37s","accounts_in_message":[],"_revision_number":3},{"id":"3c37bbf94923c8f1e7239ce6a41cf274a021eb86","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 09:47:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e9e2cc8a3378cc28edc91af388a484f5973ed081","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 09:49:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6002fed172244f86bd725c7dafd2687e : SUCCESS in 1m 00s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/945d1f3fe88f456790b7449462eaecd2 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70e365890c164c00ab9c25b54a44891c : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b89c929c12fb262f59ba320a37f2a5bf4109df98","revisions":{"f143a639521ce7ce3bde8060fcf763ea26ad21d6":{"kind":"REWORK","_number":1,"created":"2021-08-04 11:00:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/803461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/61/803461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/803461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/803461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/803461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/61/803461/1"}}},"commit":{"parents":[{"commit":"7e264e1f21ee396d046e0df0c1a4f4218039e4aa","subject":"Fix oslo policy DeprecatedRule warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e264e1f21ee396d046e0df0c1a4f4218039e4aa"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-04 10:57:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-04 10:59:13.000000000","tz":540},"subject":"Enable debug of pip-extra-reqs/pip-missing-reqs","message":"Enable debug of pip-extra-reqs/pip-missing-reqs\n\n... so that we can understand behavior of the command more in detail.\n\nChange-Id: I6ae2740557f0e62b6805e5dadd025e7bf44ec57d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f143a639521ce7ce3bde8060fcf763ea26ad21d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f143a639521ce7ce3bde8060fcf763ea26ad21d6"}]},"branch":"refs/heads/master"},"60ad1acfa53887e4e9b9d7b961c810a23274a271":{"kind":"REWORK","_number":2,"created":"2021-08-06 08:45:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/61/803461/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/61/803461/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/803461/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/803461/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/803461/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/61/803461/2"}}},"commit":{"parents":[{"commit":"7e264e1f21ee396d046e0df0c1a4f4218039e4aa","subject":"Fix oslo policy DeprecatedRule warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e264e1f21ee396d046e0df0c1a4f4218039e4aa"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-04 10:57:49.000000000","tz":540},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-08-06 08:45:46.000000000","tz":120},"subject":"Enable debug of pip-extra-reqs/pip-missing-reqs","message":"Enable debug of pip-extra-reqs/pip-missing-reqs\n\n... so that we can understand behavior of the command more in detail.\n\nChange-Id: I6ae2740557f0e62b6805e5dadd025e7bf44ec57d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60ad1acfa53887e4e9b9d7b961c810a23274a271"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60ad1acfa53887e4e9b9d7b961c810a23274a271"}]},"branch":"refs/heads/master"},"b89c929c12fb262f59ba320a37f2a5bf4109df98":{"kind":"REWORK","_number":3,"created":"2021-08-06 10:05:14.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/61/803461/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/61/803461/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/803461/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/803461/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/803461/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/61/803461/3"}}},"commit":{"parents":[{"commit":"7e264e1f21ee396d046e0df0c1a4f4218039e4aa","subject":"Fix oslo policy DeprecatedRule warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e264e1f21ee396d046e0df0c1a4f4218039e4aa"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-04 10:57:49.000000000","tz":540},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-08-06 10:04:56.000000000","tz":120},"subject":"Add caps for pip-extra-reqs/pip-missing-reqs","message":"Add caps for pip-extra-reqs/pip-missing-reqs\n\npip-check-reqs fails with latest pip, cap both packages until\nhttps://github.com/r1chardj0n3s/pip-check-reqs/issues/66 is fixed.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nChange-Id: I6ae2740557f0e62b6805e5dadd025e7bf44ec57d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b89c929c12fb262f59ba320a37f2a5bf4109df98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b89c929c12fb262f59ba320a37f2a5bf4109df98"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
