)]}'
{"id":"openstack%2Foctavia~804160","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I6ae2740557f0e62b6805e5dadd025e7bf44ec57d","project":"openstack/octavia","branch":"stable/victoria","topic":"pip-debug-stable/wallaby-stable/victoria","hashtags":[],"change_id":"I6ae2740557f0e62b6805e5dadd025e7bf44ec57d","subject":"Add caps for pip-extra-reqs/pip-missing-reqs","status":"MERGED","created":"2021-08-11 09:50:14.000000000","updated":"2021-08-12 09:54:24.000000000","submitted":"2021-08-12 09:51:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804160-pip-debug-stable/wallaby-stable/victoria","cherry_pick_of_change":804159,"cherry_pick_of_patch_set":1,"meta_rev_id":"c208345cd7fe11ac5e86b3b41acfc7b658db16f9","_number":804160,"virtual_id_number":804160,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-12 09:51:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-08-12 07:42:47.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2021-08-11 15:24:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2021-08-12 07:42:47.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-11 09:50:14.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-08-11 12:04:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-11 15:24:52.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-08-12 07:42:47.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"a63fd7b267ba0221fc27650effa21f40725e103d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-08-11 09:50:14.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"bdac36f979a37639e5c536375dc41c038e3c9024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 12:04:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bff00efb90994334897d5feb3280d152 : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efc7c711fe824a32964b0b9212505b3c : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adbdca2dd46e4226890bc9fefdcbb2bb : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6c15b31d0844285ab1aaf363b284798 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dae5c22fed148dd8bcc527961a5c867 : SUCCESS in 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cbbf70d9aef4f1da62b7781d85b9f0f : SUCCESS in 5m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0103a714a50a4fcf8eec9859f345c97b : SUCCESS in 6m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6de75da671cf481cb8abbf650ebab3bc : SUCCESS in 8m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2c6a2a70bbb64cba84ca1ecc793d0e9a : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/072197bc96ab4b3cb84c362492169e9b : SUCCESS in 7m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a640cdadcbe04d18938cfc3169ed037a : SUCCESS in 1h 34m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a78e399ec6574a38901ab39dc271a384 : SUCCESS in 1h 37m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/774ef43415ee4003b85b8855d7f2c9d2 : SUCCESS in 34m 15s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/59e0a5120b044daeb4273ef1bfc3322d : SUCCESS in 32m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/213ea7ae384443a684c11c9dc903d1a4 : SUCCESS in 33m 07s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/545fbc325bb04eae968b756ab4611b43 : SUCCESS in 1h 28m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9c2af0b137444a168e41081a4608afaa : FAILURE in 27m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/853c91b28e9c45059ae5db9a7009c45d : SUCCESS in 1h 41m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5aeac2963a084cf8890d213417c46e46 : FAILURE in 2h 11m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/93e91995d16d48c6aaf92696730810b6 : FAILURE in 30m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/058f58640f3f42c1a03b62ed8809724d : FAILURE in 31m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4770cbbdb4619190d6cfeb770da93207a519f420","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-11 15:24:52.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"e48aabfeb1f320c85532a4c083ec9abe60a25611","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-08-12 07:42:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9a7e5202dffb41b410ce6c987bb36a798a4fad1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 07:43:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2bfff4421435363adc48516907b91f9bc9a30756","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 09:51:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0ed6dade8f04875bdb32b8c19f1e9df : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58f35579b1fc4f5cb4560165d2333b40 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2801922e55e546f0a2443b089f59636b : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83de4de3d20248cd9d61d990e95c1938 : SUCCESS in 12m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/890f7f0a283a43c5b78f786ee6d53f75 : SUCCESS in 6m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d4a2bff7bae64c849d6cd38b73400aac : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f0432d1f653845cf9cd5c6847d2f1742 : SUCCESS in 7m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4ebb71386a5d40b29e61b1f2d18a0abf : SUCCESS in 1h 33m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fce79e37fe63431aa1a70616af547d14 : SUCCESS in 2h 07m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e01253d105ec4979805e07bb1861be0f : SUCCESS in 43m 23s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5b78871f9e6245b98962ef3b81f3b68d : SUCCESS in 46m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02739fc198844ce78dc9e700a319e6e8 : SUCCESS in 44m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1074c5fe7b274385a3e5c369f804a2fa : SUCCESS in 1h 30m 39s","accounts_in_message":[],"_revision_number":1},{"id":"681c966b83054798cd18f8666f8f8bb3b4a06bd0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 09:51:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c208345cd7fe11ac5e86b3b41acfc7b658db16f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 09:54:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9942beb0b1d4ca0ad02754231c6c4f2 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64f59b7e69f8401197014d84fc952d39 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18413fab504d55323c9d9bc0cf71fefc0e1731d8","revisions":{"18413fab504d55323c9d9bc0cf71fefc0e1731d8":{"kind":"REWORK","_number":1,"created":"2021-08-11 09:50:14.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/60/804160/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/804160/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/804160/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/804160/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/804160/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/804160/1"}}},"commit":{"parents":[{"commit":"74c0ff2a401195c6a1f3b784b954729a45ff98f6","subject":"Enable lo interface in the amphora-haproxy netns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/74c0ff2a401195c6a1f3b784b954729a45ff98f6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-04 10:57:49.000000000","tz":540},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-08-11 09:50:14.000000000","tz":0},"subject":"Add caps for pip-extra-reqs/pip-missing-reqs","message":"Add caps for pip-extra-reqs/pip-missing-reqs\n\npip-check-reqs fails with latest pip, cap both packages until\nhttps://github.com/r1chardj0n3s/pip-check-reqs/issues/66 is fixed.\n\nCo-Authored-By: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n\nChange-Id: I6ae2740557f0e62b6805e5dadd025e7bf44ec57d\n(cherry picked from commit b89c929c12fb262f59ba320a37f2a5bf4109df98)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/18413fab504d55323c9d9bc0cf71fefc0e1731d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/18413fab504d55323c9d9bc0cf71fefc0e1731d8"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
