)]}'
{"id":"openstack%2Foctavia~804175","triplet_id":"openstack%2Foctavia~master~I681f06217baf08018462a8210ba595ecb786b281","project":"openstack/octavia","branch":"master","topic":"config-generator","hashtags":[],"change_id":"I681f06217baf08018462a8210ba595ecb786b281","subject":"Regenerate octavia.conf","status":"ABANDONED","created":"2021-08-11 00:13:02.000000000","updated":"2022-06-21 23:08:54.000000000","total_comment_count":12,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"4827348926afcb921681d356d33ce17aa72567d5","_number":804175,"virtual_id_number":804175,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-12 10:17:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":-1,"date":"2021-09-14 22:07:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-11 00:25:38.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-08-11 01:59:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59aac4729a882bac6ef91b1b94eace43067633c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-11 00:13:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"522f71e7fee79d7e27fcff45df17549f0e91ea83","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-11 00:25:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nIn general I am good with updating this file and pulling in more of the third party library settings.\nHowever, there is a reason we do not automatically generate this file using the oslo config tools.\n\nThere is a lot of comments, and clarifications in this file that are not present in all of the oslo config comments.\n\nWe are going to have to take a very careful look at what this changes and clean it up.","accounts_in_message":[],"_revision_number":1},{"id":"63d725fbe6ebc42a8721ac67499d2f93779ff685","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-11 00:27:15.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThis is just a few obvious comments and not a full review. I will have to allocate an hour or so to review this line by line.","accounts_in_message":[],"_revision_number":1},{"id":"ef8087253d9c03f6e9709dd607c30d879ab10051","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-11 00:43:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"988d5fb9f8d15937c6fa375c9a1b60fc08fdb2e9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-11 00:45:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a49ab7cefd0b98a027f40082c13a0a3d07960696","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 01:59:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/632ad670f37a40b3af354221e427b68d : SUCCESS in 12m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b58645a2b1d744a4bb8784328109157a : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f413ab2f074416d88ac92a5681e93af : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd9a866aa2c641e8b30c7736cf7e04f5 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9af7e5245cb483288ecaefef1a0e45c : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e33548d0fb574b599a4d7783d87dbf39 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/164807c115944c9ca1960439e54851a7 : SUCCESS in 12m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bbe46d2ec95446e68bd55c354a3f0796 : SUCCESS in 5m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b1a6007bba5e4b45b9a4dca3cd247637 : SUCCESS in 9m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/50c193d8f2c14df0af52f564944b214c : SUCCESS in 1h 41m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7b9bd0621a2e45fb9042ac796dbf7f64 : SUCCESS in 1h 18m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/670d4b3eceed4588b210ce830c736bca : SUCCESS in 36m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/695fc9fbbc7d4ef08633e03bccb1f36c : SUCCESS in 31m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/10cc27937e6340a38a9d9018036a8ff2 : SUCCESS in 47m 48s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f7a685479c9d4fb5b94359192a82d8d6 : SUCCESS in 1h 27m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bb652c77a0a24a99855016cee9be5c87 : FAILURE in 14m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cfc9da0777b34999860f9ed8589a1848 : SUCCESS in 1h 24m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/160eac19c9a14098984dbe2a50a47a32 : SUCCESS in 1h 23m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/aae90d32650c499a953e9960501724cf : SUCCESS in 1h 25m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/409a6f9d4e2b43838f5ac93c56db9201 : FAILURE in 11m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"583d39782059f47fb4091723a027e561cfdf67fb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-11 05:54:03.000000000","message":"Patch Set 1: Workflow-1\n\nThank you Michale for you attention.\n\nJust FYI. The reason I started working on this was that I noticed currently RDO relies on octavia.conf in Octavia repo which doesn\u0027t included good explanation about quota parameters. The current goald is to let distro like rdo use config-generator to provide a base .conf file.","accounts_in_message":[],"_revision_number":1},{"id":"ac6f0bfdfc5c7cf9537581ac3d0fc573768b2c81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-11 05:56:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fabf9fe70c70d888093c9f6fbb4ac04c063efa59","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-11 05:57:47.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4692d9c743123af8368b7f4a663dfc6f965b193b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-11 06:03:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d3837793a138eea662e6a0948628cef9add88db4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-11 06:12:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a23d8088af726cd1a01f12c68296f8971c0a4474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 08:12:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f7ad416a7214306a44b40f80aa04d07 : SUCCESS in 14m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85f29a1bd0ce47468139d14be2ac3517 : SUCCESS in 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54c3613cc78448408660e4e74f7c3005 : SUCCESS in 8m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2d0087cc4b84c58a4248a9b2f859c50 : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/822b99fe35474687990488f364a43f77 : SUCCESS in 9m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3434a7707e06421d9a410bff21d788c5 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da399b66bf4b4073bc6ee467625d4b1e : SUCCESS in 13m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/97e5db2ba4a8471a81f76365db7a8ca9 : SUCCESS in 8m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/13454a7adeda44e995425ba8b4390694 : SUCCESS in 11m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f1bd58c4c49b4fcfa15da64e397a2696 : FAILURE in 9m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ddaab17633cc45889fb514f47835b264 : SUCCESS in 10m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/02e5e7795a254b65a5c522c5e3a4846b : SUCCESS in 1h 46m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1d22f2cb9a214ce78e37916b0d5983a5 : SUCCESS in 1h 23m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4c12421799fe4273bb5d7c0e0d8178af : SUCCESS in 37m 33s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6e3dd135ab2342788ec11cbd9fccc5ab : SUCCESS in 36m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d26af9ad46484b27bf30cfa55e62e51b : SUCCESS in 50m 30s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/67294c67dc5a4ec09c13540b471d4674 : SUCCESS in 1h 23m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/197b90072d3247e692cd4c90330a7173 : FAILURE in 19m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a71393b7e8a54cf1aa367cf6bfded736 : SUCCESS in 1h 28m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/edf8688139c64f7e91ff2ca59088695e : SUCCESS in 1h 25m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/21a3a89da6a0425ebc2d8c66732e6f34 : SUCCESS in 1h 27m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d7c15d4e13a44604807ef501f6bb3745 : FAILURE in 12m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2890d097dcd9f8af7d5f5a959c2bf85e73e2fbd5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-08-11 15:20:50.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5190181d6c8054932733b05b2003f924b3303d13","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-11 15:28:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0c57e75859bfb3602c8c0b11c0ff49efc0f39e2e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-11 16:09:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"63a0ace8dea3e0449d5d1c2fed148558b474ed20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-12 01:59:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4be9a08778a8ce1ebef4bc9c4a68ca8c5069e580","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-12 02:02:30.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"11c5c2878047476cc519d7258adb3401a720bc09","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-12 02:05:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a6fea9e68e67137257d59ea8b04a3e1802f51126","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-12 02:22:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e5a3dc3d31ef6ef9f6995d8b2e3dad3d62c1da68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 04:03:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96ca3c4607c5424b923e2040b218cede : SUCCESS in 15m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e34dcee9f224e018a7d0619e614de27 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec30dec5680242d0aac599b89beac346 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/746d3b143a9740729102365d75f0fa4b : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a809e1dee4094d40ab54fec54224e729 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee060505f44f40d284001674aba7d88f : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baf406d0e1b1478eb882d2cd26c5cc4a : SUCCESS in 12m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/db9c4ba80b9a4349a264f9bb1651abfe : SUCCESS in 5m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/db27130708d545848a569ca7b6d0d257 : SUCCESS in 13m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4e3dd527d0584b88868094821b2fe14f : SUCCESS in 1h 48m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0110f059d5ce4ae7964aa7930e9dd25b : SUCCESS in 1h 34m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ce91ab07defd4f8db8a265692a769ca2 : SUCCESS in 35m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/897581d73da642c990cac731aa527a7a : SUCCESS in 40m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/08ebf0bcda7f451bbf807e02b1fc8fb7 : SUCCESS in 48m 32s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ddcafe37680c4d73b85d6263a24b9944 : SUCCESS in 1h 58m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/00136474b2f644908b079ff4030d8a16 : FAILURE in 13m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4e0e3ac322d04885be2535f247b2ac72 : SUCCESS in 1h 22m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/26a4f56df2924c4591a8c302a2bf2864 : SUCCESS in 1h 19m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c5571448f03b4250b43a435747184b86 : SUCCESS in 1h 41m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/24be2312cb554898b6d797250f0b40aa : FAILURE in 11m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"295676e27b20962b407cd6acc74349403cb7d3e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-12 08:29:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9acb5836a3743a16f01d78aa4fc28a09834712df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-12 08:32:19.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a4ef5ad181546a53d6a50230971877b9bca2bb35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 10:17:40.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bfd909f3d3243b99af0e9b9fee95f61 : SUCCESS in 11m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/946a34c209ea47509c9194a918e8aa9c : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf6dacee2cc483b8e2544c469827cc6 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce7bf248178448a5bdab909995fcdff3 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b957f870c08646398a256b520c7848bc : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f58cfc7e01c49b3a46ddc165f9510f7 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ff04c085ad9453382b4b8a6f687a129 : SUCCESS in 13m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b0cd2a243103494c927b2af8f6283936 : SUCCESS in 7m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a2b1f81c0f974b5186a011a6f3bef399 : SUCCESS in 9m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4ce15fb6a47a460589ca2e8b0eb7d1f4 : SUCCESS in 1h 40m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a62b53da6254c18b079b10333bd5ad4 : SUCCESS in 1h 14m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/66ac1e42c6d8497cb8955a6864b5ab48 : SUCCESS in 32m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2d5c95af32ab4ce2870545d5602224d6 : SUCCESS in 31m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/263d8723234944c991948c9d72983e38 : SUCCESS in 48m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/07947c6745864f41808d455776a49e24 : SUCCESS in 1h 27m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a55fd6d5ce5a403ba697be153a4c1791 : FAILURE in 16m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/32a91140f8b94389b8a31c5e8508d8c3 : SUCCESS in 1h 23m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/100d7d6a08274e8c8b6e569f1dd6afc5 : SUCCESS in 1h 38m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/601827926ca54eddb6d5a76755de6999 : SUCCESS in 1h 15m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fe5ed83d6f1d409cbdaf86f2122305d1 : FAILURE in 9m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8947941855a80f110080335fd30f36bb2065edb5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-09-14 22:07:47.000000000","message":"Patch Set 7: Code-Review-1\n\nI\u0027m not sure I see the value of this. This is replacing the annotated, managed version, with a dump from oslo config generator. If we don\u0027t want to have better formatting and more commentary about the config (like the existing etc/octavia.confg), why not remove this and just have people run the oslo config generation tool?","accounts_in_message":[],"_revision_number":7},{"id":"120748699361afa6bf15b8d5ace99738dcf6c43c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-15 10:38:24.000000000","message":"Patch Set 7:\n\n\u003e Patch Set 7: Code-Review-1\n\u003e \n\u003e I\u0027m not sure I see the value of this. This is replacing the annotated, managed version, with a dump from oslo config generator. If we don\u0027t want to have better formatting and more commentary about the config (like the existing etc/octavia.confg), why not remove this and just have people run the oslo config generation tool?\n\nI think we can remove this file once we completely fix the wrong items rendered by oslo-config-generator, although having a renfered version might be useful for quick reference. Actually some projects like glance has a rendered file which is updated each cycle.\n\nMy main intention of this patch is to show the difference between the existing file and the auto-generated one.","accounts_in_message":[],"_revision_number":7},{"id":"4827348926afcb921681d356d33ce17aa72567d5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-21 23:08:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"f190946cef6b9c60e543cacbbbbb89e4c71e6a35","revisions":{"c7b540ab74e850242e1fce6fcb27f92f6ccab8c7":{"kind":"REWORK","_number":1,"created":"2021-08-11 00:13:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/804175/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/804175/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/804175/1"}}},"commit":{"parents":[{"commit":"7e264e1f21ee396d046e0df0c1a4f4218039e4aa","subject":"Fix oslo policy DeprecatedRule warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e264e1f21ee396d046e0df0c1a4f4218039e4aa"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-11 00:08:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-11 00:11:03.000000000","tz":540},"subject":"Regenerate octavia.conf","message":"Regenerate octavia.conf\n\nThis change replaces the example octavia.conf file in octavia repo so\nthat the file includes more complete list parameters with corresponding\ndescriptions.\n\nChange-Id: I681f06217baf08018462a8210ba595ecb786b281\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7b540ab74e850242e1fce6fcb27f92f6ccab8c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7b540ab74e850242e1fce6fcb27f92f6ccab8c7"}]},"branch":"refs/heads/master"},"b94acce7f0b3ead508c553ce8b2da066bf59276a":{"kind":"REWORK","_number":2,"created":"2021-08-11 05:56:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/804175/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/804175/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/804175/2"}}},"commit":{"parents":[{"commit":"4b16373b166011e384abeaa1f78ffb9f1c8ae468","subject":"Add conf file for oslo-config-generator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b16373b166011e384abeaa1f78ffb9f1c8ae468"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-11 00:08:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-11 05:56:09.000000000","tz":540},"subject":"Regenerate octavia.conf","message":"Regenerate octavia.conf\n\nThis change replaces the example octavia.conf file in octavia repo so\nthat the file includes more complete list parameters with corresponding\ndescriptions.\n\nChange-Id: I681f06217baf08018462a8210ba595ecb786b281\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b94acce7f0b3ead508c553ce8b2da066bf59276a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b94acce7f0b3ead508c553ce8b2da066bf59276a"}]},"branch":"refs/heads/master"},"f261422dcb74edcebaaf127eb4465d106d7bf7a9":{"kind":"REWORK","_number":3,"created":"2021-08-11 06:12:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/804175/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/804175/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/804175/3"}}},"commit":{"parents":[{"commit":"4b16373b166011e384abeaa1f78ffb9f1c8ae468","subject":"Add conf file for oslo-config-generator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b16373b166011e384abeaa1f78ffb9f1c8ae468"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-11 00:08:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-11 06:10:46.000000000","tz":540},"subject":"Regenerate octavia.conf","message":"Regenerate octavia.conf\n\nThis change replaces the example octavia.conf file in octavia repo so\nthat the file includes more complete list parameters with corresponding\ndescriptions.\nAlso, description about allowed values are re-implemented using\nthe \"choices\" interface so that oslo-config-generator can render it\nproperly.\n\nChange-Id: I681f06217baf08018462a8210ba595ecb786b281\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f261422dcb74edcebaaf127eb4465d106d7bf7a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f261422dcb74edcebaaf127eb4465d106d7bf7a9"}]},"branch":"refs/heads/master"},"1d7fd97d7de4d9228f8bd891866528f2874d6ec4":{"kind":"REWORK","_number":4,"created":"2021-08-12 01:59:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/804175/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/804175/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/804175/4"}}},"commit":{"parents":[{"commit":"bce74bbc556c22e7f3b9e7c63f3b34665aa225d6","subject":"DNM: Sort parameter sections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bce74bbc556c22e7f3b9e7c63f3b34665aa225d6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-12 01:57:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-12 01:59:15.000000000","tz":540},"subject":"Regenerate octavia.conf","message":"Regenerate octavia.conf\n\nThis change replaces the example octavia.conf file in octavia repo so\nthat the file includes more complete list parameters with corresponding\ndescriptions.\nAlso, description about allowed values are re-implemented using\nthe \"choices\" interface so that oslo-config-generator can render it\nproperly.\n\nChange-Id: I681f06217baf08018462a8210ba595ecb786b281\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1d7fd97d7de4d9228f8bd891866528f2874d6ec4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1d7fd97d7de4d9228f8bd891866528f2874d6ec4"}]},"branch":"refs/heads/master"},"84ddff7660a129d2226e74334e0f17bcd63e3648":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-08-12 02:02:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/804175/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/804175/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/804175/5"}}},"commit":{"parents":[{"commit":"bce74bbc556c22e7f3b9e7c63f3b34665aa225d6","subject":"DNM: Sort parameter sections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bce74bbc556c22e7f3b9e7c63f3b34665aa225d6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-12 01:57:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-12 02:02:16.000000000","tz":540},"subject":"Regenerate octavia.conf","message":"Regenerate octavia.conf\n\nThis change replaces the example octavia.conf file in octavia repo so\nthat the file includes more complete list parameters with corresponding\ndescriptions.\n\nChange-Id: I681f06217baf08018462a8210ba595ecb786b281\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84ddff7660a129d2226e74334e0f17bcd63e3648"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84ddff7660a129d2226e74334e0f17bcd63e3648"}]},"branch":"refs/heads/master"},"5342ad0a1c095f28aebe5e5409241ef842444312":{"kind":"REWORK","_number":6,"created":"2021-08-12 08:29:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/804175/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/804175/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/804175/6"}}},"commit":{"parents":[{"commit":"e4415c9dc4e8093fec258bcb40d958b2e51838de","subject":"DNM: Sort parameter sections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4415c9dc4e8093fec258bcb40d958b2e51838de"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-12 01:57:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-12 08:28:08.000000000","tz":540},"subject":"Regenerate octavia.conf","message":"Regenerate octavia.conf\n\nThis change replaces the example octavia.conf file in octavia repo so\nthat the file includes more complete list parameters with corresponding\ndescriptions.\n\nChange-Id: I681f06217baf08018462a8210ba595ecb786b281\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5342ad0a1c095f28aebe5e5409241ef842444312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5342ad0a1c095f28aebe5e5409241ef842444312"}]},"branch":"refs/heads/master"},"f190946cef6b9c60e543cacbbbbb89e4c71e6a35":{"kind":"REWORK","_number":7,"created":"2021-08-12 08:32:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/804175/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/804175/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/804175/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/804175/7"}}},"commit":{"parents":[{"commit":"44a2a287993c6e0b636ed905103dd44800ce0c6b","subject":"DNM: Sort parameter sections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/44a2a287993c6e0b636ed905103dd44800ce0c6b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-12 01:57:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-12 08:32:09.000000000","tz":540},"subject":"Regenerate octavia.conf","message":"Regenerate octavia.conf\n\nThis change replaces the example octavia.conf file in octavia repo so\nthat the file includes more complete list parameters with corresponding\ndescriptions.\n\nChange-Id: I681f06217baf08018462a8210ba595ecb786b281\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f190946cef6b9c60e543cacbbbbb89e4c71e6a35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f190946cef6b9c60e543cacbbbbb89e4c71e6a35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
