)]}'
{"id":"openstack%2Foctavia~804180","triplet_id":"openstack%2Foctavia~master~I619b2ce6912764e5a76d5eb41935cbcebd2919b5","project":"openstack/octavia","branch":"master","topic":"config-generator","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-07 16:22:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I619b2ce6912764e5a76d5eb41935cbcebd2919b5","subject":"Add missing [cinder] parameters","status":"MERGED","created":"2021-08-11 00:43:06.000000000","updated":"2022-07-07 16:24:04.000000000","submitted":"2022-07-07 16:22:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804180-config-generator","meta_rev_id":"ef61f9deed4d478da83853db2638aeddae710642","_number":804180,"virtual_id_number":804180,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-07 16:22:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"all":[{"value":1,"date":"2022-06-21 23:40:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2022-06-22 02:14:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-21 23:40:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2022-07-07 14:05:44.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2022-07-07 14:05:44.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-11 02:35:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-31 09:06:20.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-06-21 12:09:04.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-06-21 13:22:30.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2022-06-22 02:14:41.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"}],"messages":[{"id":"391429ef11db2a73d8710320604f0d9e7e8a9711","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-11 00:43:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d359ec0c04f5bbadace39c912f3ccbbd5468ba79","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-11 00:47:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"769ee2a7f536e487ab06672ac859987efee9a8c3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-11 00:47:32.000000000","message":"Topic set to config-generator","accounts_in_message":[],"_revision_number":1},{"id":"6dae608c07a5b1bb2b030518b1bc955eb010d562","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 02:35:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdbef6c48ca9495ba87ea51f7ead7c76 : SUCCESS in 12m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9960bd386de24423a775333355134d98 : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afe1b176d4b543fc868ca0d012514755 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05967c0eefd940a6911bef8f827173bc : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27a94514e5df4a16828c542a77a40342 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc8f87c9e7d14ef989d6443df90a6479 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5116692be5c42819600e9dd7dddd0c2 : SUCCESS in 14m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e48cfb8312cc48a5a677de9fb2413be0 : SUCCESS in 6m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b51e36a4561240bb9902f38c43721c26 : SUCCESS in 11m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/47fe6bd76c974c1c9ae37147d2b56edb : FAILURE in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8b0fa45cdd8f43a1a891af26d49b56f2 : SUCCESS in 9m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0b25039b7ec74b99b34a321ea544d63a : SUCCESS in 1h 37m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/45b4961a65e944bf912457e29a15120a : SUCCESS in 1h 35m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/94fbf3492ec449299d40f2a364018b27 : SUCCESS in 39m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da1414157b674808929ef57f568114e3 : SUCCESS in 40m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6d229ab2ed3747ca8edb8893525d9791 : SUCCESS in 48m 04s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/658333d75e344514ad7076d8c5e04902 : SUCCESS in 1h 42m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1845bad8426b4746bf1c38b8b4864579 : FAILURE in 18m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f478de69182747d98d2aa4329af4ca84 : SUCCESS in 1h 42m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0fb36e56e7ab4e0f889ecd7e411210fb : FAILURE in 15m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/677d01c104e041b988ebacbbdec72beb : SUCCESS in 1h 50m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a28d9ae742ff44e9a69b9d835049116a : FAILURE in 8m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"74a71cd4574699cf937c732ae8d280527144ec98","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-11 10:00:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"10362be61ffe33d04ef4c7425829b69833aef072","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 11:54:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65d447a2b79f42c6bd0466d4a79eb5be : SUCCESS in 11m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa942b372dc74614921f3bc661752cfd : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a7506f1a98940a5a89dc8282e61e498 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1b7e20a145147fbb1abe0c2dc0e76e6 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff7be7c46c0540c7a00fcfd4ab51347c : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbb7d55ec1bc479cbf551b0431fa46aa : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87b2c7c8b7fd469db5793aca374deacd : SUCCESS in 13m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/043526f363a14979bafa902bede9094d : SUCCESS in 6m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/eb5bd2d798a34afd9970072deac87eec : SUCCESS in 10m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/274cd3c4688249e8b557338df9b57765 : SUCCESS in 5m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/acadc9736a11444f9aeb3c6087889009 : SUCCESS in 8m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/525256465ea5416282020dbfecbac7fd : SUCCESS in 1h 39m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c45d619aceb14858a45a5e55d81567d5 : SUCCESS in 1h 21m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/692338cf9d9e40ee9d1c0954a8e3d142 : SUCCESS in 34m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a6eb39da43ca47a28181a87f1bd016a6 : SUCCESS in 35m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d979b4ad5c344ee48ec6aaa9f886c18f : SUCCESS in 48m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0c434c844edc4eb9b08b8e2d035f4b2e : SUCCESS in 1h 15m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d1efd2ad019947a58503d0bba3261f82 : FAILURE in 14m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/98c38d60eeaf44b99b88ed9dd9a6c5ff : SUCCESS in 1h 34m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0a8df180d9564d0aaf280b239ac6e0a8 : SUCCESS in 1h 52m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7fbca16f33b246ad856558e80695b5f7 : SUCCESS in 1h 35m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/76af2cfeff254ae1b157e8ecf6c34599 : FAILURE in 9m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e639ed87ad5513fd8a67455673958e52435d558","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-12 01:58:23.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e4d96502e8a0e24bedd6b05e0a287f5af0553864","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 03:52:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8110bf21d2d40c9b8b406a12856ea91 : SUCCESS in 11m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d92203aa42a4ca3a45b312810eeff22 : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddcc532648df4701aebbaababa208c67 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa33dd6560de491f9a235a0c7b1bda02 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9815ea1c35214b939409cab886ab1052 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26e7563426074f469131c87895eb1d0e : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/734aab1921ff46de8ff31c51ec1bed11 : SUCCESS in 12m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/85d2a6e0d2624874863edcc83e4e1da5 : SUCCESS in 6m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d59d390ecc4b42a087884802bcab6639 : SUCCESS in 10m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8045a703315a4194ab0845babe054cb5 : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3052e1577c97421386f15e58e20c356d : SUCCESS in 8m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6a0636e4193f43c597c3f4c0fdeda2c4 : SUCCESS in 1h 39m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b22b9915de92423bb6d01a2b72b83ef0 : SUCCESS in 1h 28m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/524413fe55e74c42af839b23344b6d75 : SUCCESS in 37m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4fd24cbaf00441229ae525dc60556327 : SUCCESS in 40m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8284c1802d154b58bc9a38088233cc82 : SUCCESS in 1h 05m 46s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6982c820f4034f5f94ed4b9bf8e2dbac : SUCCESS in 1h 40m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b2c52576ed2a4ffaa263dda30d8e42a0 : FAILURE in 17m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ae31bdb1fb354c4d95e8d981581a50f4 : SUCCESS in 1h 48m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/249706b860c94ff3a1d462e2fda477bf : SUCCESS in 1h 39m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/72badd11dd614fee9817843a6ab8e7c1 : SUCCESS in 1h 49m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7e9cf7626c5345d99cca07828a87a9a1 : FAILURE in 10m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c7600c0bf4e49cadfdcdd63de3a8476b6c1ca226","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-16 06:08:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b1cf0270bd1bf50ce7a9408ab9f051b7944d846","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-31 09:06:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3c8fb7c5e43edb4936540b5d74bc8207317396ac","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-31 09:07:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ebcc5ac8e73e9bc5d06ba2bf5f5bd1f775fbe108","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 11:21:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cb35667e3d44d0981678034ef50b3ec : SUCCESS in 10m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8326a7b1610744ac92b73789cc26d262 : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/174ce7c966cc48138cd104fb0eda63cd : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec90b3dd31fb45d792793619439a5cc7 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f661f3e064a34b4f8f01c8ac3204bf40 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45de245ce7f54aefa6ab12be15e266a3 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8846eaf6c04b4b50b6d894bc70522f91 : SUCCESS in 12m 54s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f9edbf69b8794ff4ba90b13312475419 : SUCCESS in 6m 12s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b143074ff3744325bd9053a99b2a4511 : SUCCESS in 9m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a6b488cb13144a73a909f3b17a609974 : SUCCESS in 6m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d993aa7b08224d899b581d468ac15ce3 : SUCCESS in 9m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/94445b9665204ee7bdc3f3f5e10e5454 : SUCCESS in 8m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bf0bbb0d550f4fa9b8bb378ecee96077 : SUCCESS in 7m 39s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4e0c661d575648e992e4927bd33e5910 : SUCCESS in 2h 10m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2da18f4804d4e9f97bd247ce5ac7ff7 : SUCCESS in 1h 21m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eb80637d71b84233a38fbd5228718333 : SUCCESS in 34m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/09fdda4ed32849e7bb86908379a882df : SUCCESS in 51m 15s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3fa3e440b6724550972f1fcfff5e3f89 : FAILURE in 12m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53d387689f0d4ec3b851040234208318 : SUCCESS in 36m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/39d01ff221014ec1999b0f28dd987bee : SUCCESS in 1h 37m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2b4af46d3578415485b4c2b48ca4657f : SUCCESS in 1h 57m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/69640fae73c64878a8c7b1768a26b427 : SUCCESS in 1h 40m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/094516e8c6534269a13c4713cc450687 : SUCCESS in 1h 27m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/873a603733424c3db2f0759ff26ddbf5 : SUCCESS in 1h 23m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/f447af0d0de9447eac867a90098e5953 : FAILURE in 10m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dff9f8efaacc5cec2fc6dbd291aedae056df5c9e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-06-21 12:09:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f4dc180fab4c5ab0987b6f00458230eb0f597bb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-21 12:18:16.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"add7e0645494c572afc2db39e45132346ab0de35","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2022-06-21 13:22:30.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"00d707ffeaa04033e05c7fbdf4ccf4f333ac0491","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 14:17:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba3693d6631f4dafa4b383c0b25b2b03 : SUCCESS in 10m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee43fb7c462f497e8b9aa4fad0c63940 : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b34c0841990424fa7086b98f1309ac5 : SUCCESS in 7m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8b15316cc6047beba64a51ee86104b7 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf95d3d6f2c64da3a9a941526285971a : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58ba5eae27af49d59b09570fa51be907 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71d63b464e0c418ab30e1bab10ce4b95 : SUCCESS in 17m 27s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5e3158b59f58491f8492c7fa3f42baa8 : SUCCESS in 6m 52s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2894b54462e84adb845bd63ddd6624e9 : SUCCESS in 11m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a13d2b5d77ee478a8725b01b70d8f6eb : SUCCESS in 7m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/53e5b20bdd6246509f3247553b106484 : SUCCESS in 9m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f21ab1867bfa42d7b1731dee5b5bde02 : SUCCESS in 8m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9339233340a64fa39afd5acfefbabb36 : SUCCESS in 8m 36s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/efdeb745932d4baf9bd80a6e126b778d : SUCCESS in 1h 53m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ea91d533c56c40f2a8412947a7dcc8ff : SUCCESS in 1h 30m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/62195b2f2e3e47749e9170ef378749cc : SUCCESS in 39m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0b1fbf9b14b247b2927b03c42f4e4c3a : SUCCESS in 51m 37s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/baea14e157194cdba61d30253110e775 : RETRY_LIMIT in 9m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afb562c26a054ee682f386b856f1d912 : SUCCESS in 36m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a19ba750e55842e3a88916b1095e2eb3 : SUCCESS in 1h 39m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dff96bc6ee3845b9a0aee22607e680eb : SUCCESS in 1h 54m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/31685025896f49b28c5139fb5df0233f : SUCCESS in 1h 40m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/53f7890397d1404c989f5048b48d09d2 : SUCCESS in 1h 47m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ccd705912127425f8ff726c0ad0fcf65 : SUCCESS in 1h 42m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/e80e796e068e47eb8c9358bd598bbb99 : FAILURE in 12m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"903d5a4e27b6d56a241deb84db396f0cb0e9480a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-21 23:40:17.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"2496e1de76a750e08b823ab1bb33ca9256a79f51","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2022-06-22 02:14:41.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"659dac3024846f4246d7999a1d00521bad77da86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 02:22:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9b2baeeb62a46509c328fd51a6a2686 : SUCCESS in 10m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb3221cda45a41c485445c5bd118ce05 : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed936fe5a5a54877bf2fc7bb177273f3 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecb75cc40c414a15b51421cbead7db41 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75d65287e4e6476283191a95eee503e5 : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bde3fc11dcaa46e6b9fd69585c148543 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11af36e04b0742839cc252b717531f88 : SUCCESS in 16m 36s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8589fb599130465cbeb8298661704b45 : SUCCESS in 6m 42s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5fa3aef7dbeb4ca9beb55a3413cdcd7f : SUCCESS in 9m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8d97775b9f1142df95ae7d40e43b8b8b : SUCCESS in 5m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cf8ebce42fb147378565aeedcd202c56 : SUCCESS in 8m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9f84ae1af2274dc3b6e8fa5ff2fd0d36 : SUCCESS in 8m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9019a4824bf444d79bcd2f56dcb3b972 : SUCCESS in 7m 58s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e3314d9af64340c88cd96a2967c938a3 : SUCCESS in 2h 35m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e56f44365cfb4a83b55483b442494199 : SUCCESS in 1h 36m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d523590e65d443c6bff5b645fce21371 : SUCCESS in 41m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b9d537c545694f6ba876e79695b9b0e0 : SUCCESS in 52m 13s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1fecc75699da435abc14e6ef0bf96adf : RETRY_LIMIT in 8m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec9d4c52c21c4f82935006ece1828011 : SUCCESS in 35m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/15ccfc6221a145a495d9335053ade93b : SUCCESS in 1h 33m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aaa84d577e5f4590afea875c082470ca : SUCCESS in 2h 08m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c6d2ecdd6c6a4a199afb8f2b679c7f3f : SUCCESS in 1h 34m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cdcb3f10220247d6ad6b09308f4b28df : SUCCESS in 1h 33m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/94da7c75ee294ee4bab1e5e8f9bc3681 : SUCCESS in 1h 33m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/77977d221ed84ff4b84247d556579cc9 : FAILURE in 11m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b8ddcce0d0facd522514007dccd5380511964418","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-07 14:05:44.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d5e27b43896dfdae08385bbb66a9d2be261bba0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 14:06:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"db65cff38b8d762ae80157b7fa3a9ccec994e17c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 16:22:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c8998ff65784002b198de7c3278db1c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbfc7a0421334cab9a04119c6312d329 : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a2080c6d18342159404352c56862063 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb15808b9c354325970d804e67a04fcd : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32238b626ed9451a945fa8a72dcd38cd : SUCCESS in 11m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1501893a54aa4f50989d691331592079 : SUCCESS in 5m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7ca5433e820241e8b96df04039076933 : SUCCESS in 7m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d915ec0609644fc1a7b6a058d8953f65 : SUCCESS in 2h 11m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5d62dfb83bb4ab18c25118f9c379d91 : SUCCESS in 1h 58m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3cda627957564650a7d6596dfd2de9d9 : SUCCESS in 44m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a7761fd63a13469fb597b75d0c8a86be : SUCCESS in 1h 15m 16s","accounts_in_message":[],"_revision_number":4},{"id":"ad21f30b7463da85aa8d03cdd6b700e2e459f3b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 16:22:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ef61f9deed4d478da83853db2638aeddae710642","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 16:24:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d0b6c9e61ce4327b3b1133a749fc9ab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bbe215583a84a56bfc80d570673fb01 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c174312456b83ca0afc42558a13aacad82a2a353","revisions":{"ef6629cb3ed8b26fb6c02a57135babe22271cbf6":{"kind":"REWORK","_number":1,"created":"2021-08-11 00:43:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/804180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/804180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/804180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/804180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/804180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/804180/1"}}},"commit":{"parents":[{"commit":"7e264e1f21ee396d046e0df0c1a4f4218039e4aa","subject":"Fix oslo policy DeprecatedRule warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e264e1f21ee396d046e0df0c1a4f4218039e4aa"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-11 00:41:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-11 00:42:51.000000000","tz":540},"subject":"Add missing [cinder] parameters","message":"Add missing [cinder] parameters\n\n... so that the section is rendered into octavia.conf.example by\noslo-config-generator.\n\nChange-Id: I619b2ce6912764e5a76d5eb41935cbcebd2919b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef6629cb3ed8b26fb6c02a57135babe22271cbf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef6629cb3ed8b26fb6c02a57135babe22271cbf6"}]},"branch":"refs/heads/master"},"573df5b095fe178f857018b75ba93d0e38818db8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-08-12 01:58:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/804180/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/804180/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/804180/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/804180/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/804180/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/804180/2"}}},"commit":{"parents":[{"commit":"b5c9f3a7b0ddfd17b45b02b4071e95f6a5f13aa2","subject":"Exclude invalid TLS version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b5c9f3a7b0ddfd17b45b02b4071e95f6a5f13aa2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-11 00:41:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-12 01:56:44.000000000","tz":540},"subject":"Add missing [cinder] parameters","message":"Add missing [cinder] parameters\n\n... so that the section is rendered into octavia.conf.example by\noslo-config-generator.\n\nChange-Id: I619b2ce6912764e5a76d5eb41935cbcebd2919b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/573df5b095fe178f857018b75ba93d0e38818db8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/573df5b095fe178f857018b75ba93d0e38818db8"}]},"branch":"refs/heads/master"},"7897a0400abf6767d83361f287d3e636ca113051":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-06-21 12:18:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/804180/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/804180/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/804180/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/804180/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/804180/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/804180/3"}}},"commit":{"parents":[{"commit":"0059f05aeb8b7bd1fb83d45148a9926d9dafaae0","subject":"Exclude invalid TLS version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0059f05aeb8b7bd1fb83d45148a9926d9dafaae0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-11 00:41:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-21 12:18:16.000000000","tz":0},"subject":"Add missing [cinder] parameters","message":"Add missing [cinder] parameters\n\n... so that the section is rendered into octavia.conf.example by\noslo-config-generator.\n\nChange-Id: I619b2ce6912764e5a76d5eb41935cbcebd2919b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7897a0400abf6767d83361f287d3e636ca113051"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7897a0400abf6767d83361f287d3e636ca113051"}]},"branch":"refs/heads/master","description":"Rebase"},"c174312456b83ca0afc42558a13aacad82a2a353":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-06-21 23:40:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/804180/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/804180/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/804180/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/804180/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/804180/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/804180/4"}}},"commit":{"parents":[{"commit":"f166bb1481fb58acd7e7329ec394f43249001718","subject":"Fix duplicate object error messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f166bb1481fb58acd7e7329ec394f43249001718"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-11 00:41:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-21 23:39:45.000000000","tz":540},"subject":"Add missing [cinder] parameters","message":"Add missing [cinder] parameters\n\n... so that the section is rendered into octavia.conf.example by\noslo-config-generator.\n\nChange-Id: I619b2ce6912764e5a76d5eb41935cbcebd2919b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c174312456b83ca0afc42558a13aacad82a2a353"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c174312456b83ca0afc42558a13aacad82a2a353"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
