)]}'
{"id":"openstack%2Foctavia~804362","triplet_id":"openstack%2Foctavia~master~I45fd838fd5877a7da8845c100f54d2d086112c71","project":"openstack/octavia","branch":"master","topic":"config-generator","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-31 22:03:42.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-08-31 22:03:42.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-08-31 19:33:33.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I45fd838fd5877a7da8845c100f54d2d086112c71","subject":"Document available choices using oslo.config api","status":"MERGED","created":"2021-08-12 08:29:39.000000000","updated":"2022-08-31 22:04:59.000000000","submitted":"2022-08-31 22:03:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804362-config-generator","meta_rev_id":"7c5a87404bad0d7121ffcf15e5b02db4964b1048","_number":804362,"virtual_id_number":804362,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-31 22:03:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2022-08-31 14:39:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2022-08-31 19:33:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-07 15:37:39.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2022-08-31 19:33:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-12 10:45:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-22 13:41:39.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-08-21 17:29:04.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-08-21 17:29:04.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-08-31 14:39:08.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"d21f52e3b601bbb87f45a2b6019bb124ce440422","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-12 08:29:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11643bcfa42659afa6df27fba4a9817bcb260863","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-12 08:32:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1fe7dc2f750d58a10bfd38478db489dc4e1cbec2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 10:45:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d12aadb5eea5432da2ededeb6c18138a : SUCCESS in 11m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b6547967846449799722f6f287fa213 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71300214310d41bca38cb2b9f50ba9db : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2807693ba8224e70bdc9cc25d2b819ae : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/521a36fa6f9b451994b758a01b5d8c2c : SUCCESS in 8m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0921f7db4afb4fcf9ecdb673dcd769db : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/263451aeb0f5460da3b75d0bb97d72fe : SUCCESS in 14m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b3bd63b32a154d00b65221e2dbd8f84a : SUCCESS in 7m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d8e8624d6cde4e6cb073eebf7717b877 : SUCCESS in 9m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dd18e29746c64187bc968cb2f35e7c11 : SUCCESS in 6m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7b7c7504fcc144878e3d31a6e2f8d2c2 : SUCCESS in 10m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/58af357d93b745f992a4c4e560e24765 : SUCCESS in 1h 43m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f26746f6f8c642b89c62474ffb15e0d2 : FAILURE in 36m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/31132eb655f34f2c9dac9cd293bf1baf : SUCCESS in 36m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/988014d916aa4454bcc8ee46dad4acbe : SUCCESS in 31m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/101e333de2e541caa97b2b1b0c23ddc9 : SUCCESS in 1h 07m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/363692bdd23a4aa68d40c6855735bda7 : SUCCESS in 2h 11m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d86b9ea0fad04ccf91a5bf8dcf70d8e9 : FAILURE in 16m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f3cd08480432445c883bb5e756cd8845 : SUCCESS in 1h 28m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c2b522f2a833476e91ef68e70ff349e7 : SUCCESS in 1h 58m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e2291a5a32524cedb1fb6bccae03e77e : SUCCESS in 2h 10m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4c362aa3f076456586211705da8a7e45 : FAILURE in 10m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab0ab3dd6da2b80667bd6299002e6c81d90bde5d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-12 13:06:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a28d2325544d1c865e631c7af6a458565b22a984","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 15:11:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6366b1f40dcf436cbee69c9b030b9595 : SUCCESS in 11m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f12a80c5ab634f1282da694ed4e46abb : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98cfa81684de4f688815ace4e1ebdae4 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5be656df934a46b7a0c298082f127158 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe3a44e3824341b0a067df63f4d1f4c0 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30ea14168ffb48b497b6820a1650c2a6 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c11dbf933d57451bb45a20701d20c465 : SUCCESS in 12m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a8afdaa189f741ddad60b1d2a9dfc9ff : SUCCESS in 6m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4502250ee32e4abfaf923a60291c6eec : SUCCESS in 11m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/60a2f4da4cc446db959b6fc73fa92da3 : SUCCESS in 5m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9fb45d54ca2248a2a671e003ce2b143d : SUCCESS in 9m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b44576cc7b4c42bab9e64e20fac504d7 : SUCCESS in 1h 41m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/17c78ce67b6a4e94a2fbf328713d4319 : SUCCESS in 2h 00m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a89ab7c0db2f42de84bf364f5243b572 : SUCCESS in 37m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/52eb6f724a6147af89a42c9c6e9d88be : SUCCESS in 37m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5b99bbef11104685a1da8bc88c07eaa1 : SUCCESS in 1h 06m 44s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/16f6186848914954811f3925d2273a6e : SUCCESS in 1h 24m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/87dc576035fb4a7187072856d2319acb : FAILURE in 17m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1e2d671b8c2c4bf58c62b02d8c5b1b1c : SUCCESS in 1h 35m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2c22d586cc89459895ee5c5b4128e303 : SUCCESS in 2h 02m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/43183875fe63495bb7c440034972ad1d : SUCCESS in 1h 29m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b06d512573bf415b83b9d90bbe8b487a : FAILURE in 12m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"da5ed75e19988f60f03433a464bf2b034a68b0e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-21 23:09:19.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"35208a3c84677d62ec65e6d94f77588b249bc0c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-21 23:40:17.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"da076af9e570d661f526eb6392814cfd35a3195a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 02:25:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92354247a52f4ac7a2e07f297c8f3598 : SUCCESS in 10m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41859a7c7cd74c068254f89a3271a97a : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d96bd36538ae4779b31d49a451826269 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94d6522470be4a7fb179e249c51f3dd4 : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/877aca39795444a3b8ed2f4ba78624b8 : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/156bf09aac95477e96b939a3e761e856 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e03de0b40d44e4ca232e7894a119db0 : SUCCESS in 12m 04s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e77618a8040f4883b61c32dd1772cccc : SUCCESS in 6m 45s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7007a0a15fe541d6914790d7ec656a21 : SUCCESS in 9m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bf0badb07bfa4232aef7e3998932c805 : SUCCESS in 5m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d7f40fbe9f7f472ca09b72d86b469211 : SUCCESS in 9m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b66d1486806b4029866528344c698803 : SUCCESS in 8m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4a413a01b7444cd78dd3cf6b5c758925 : SUCCESS in 8m 36s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e630dcb247f04431bee44c5a678b6602 : SUCCESS in 2h 34m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d6e4d5ae0e1442cd85e76f518bbd3af3 : SUCCESS in 1h 20m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/de312887eda2421bb21bf7ba0f64a3ee : SUCCESS in 35m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5d60c87ee4674ff3becd36142bf81539 : SUCCESS in 52m 50s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/599769c972044c5ba7dbf0b52a21f62c : RETRY_LIMIT in 11m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f6cdee6897545d6b53f078a9057341b : SUCCESS in 30m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b12780114a674d59bacc5e205fa5f97d : SUCCESS in 1h 55m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/60bd8d0f18cf4aca8eed83311d0a62d9 : SUCCESS in 2h 10m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c1d290274d0f41d5882a60f789890584 : SUCCESS in 1h 44m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cc58574f51bc4a04ba6a3ce4ab0f780a : SUCCESS in 1h 42m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f0975382acc841bfa82701e936d82afd : SUCCESS in 1h 24m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/e19bcc3bb92c4a7db16ca383375fc28e : FAILURE in 12m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"24e545289f6f76130992cdc9cc2c016ca31dd32c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-06-22 13:41:39.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"59e13f64cb8214650750702ec018aca04636669b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-07 15:37:39.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9cff06db321d044576e84bfd9a0ceb8a7347cf94","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-21 17:29:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1ca1455e48b9d44f615ef1ebd811dec35bdabe09","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-08-31 14:39:08.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a1fb4c632e18258aba7d093275eeaab1730f046d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-31 19:33:33.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"767cf0d39316b3e55252e3290b699fcbfcee5a09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 19:34:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"24d2f9013353eba8c2e6ae9ce3925f5b91a57e8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 22:03:41.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/023d7166fbf8454abcc73bd4aec4ed16\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aacd641896f4521b20397ed7282b093 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10f8ddf16eeb4b328b962b81607e88fa : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/441d9e44cde042779595c80fb8c69d12 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7bb815e3dd749e9811c530a8e79e9a8 : SUCCESS in 12m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/af1720596c5d40909d11a0c94c3bb65a : SUCCESS in 5m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8d262e8578a4480180671f1a556eccf5 : SUCCESS in 8m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/037284d58c5848619f8d9ad67ef7d7ec : SUCCESS in 2h 25m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf80a754ef844028894e837e39bf6173 : SUCCESS in 1h 33m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/86d3deb2a5b141b584d8938b32a7380d : SUCCESS in 43m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8ade461214e741aebd0708a2923854fc : SUCCESS in 1h 03m 18s","accounts_in_message":[],"_revision_number":4},{"id":"09c292fc1df2b69456cd2caea81eb1a1fe71c77f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 22:03:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"7c5a87404bad0d7121ffcf15e5b02db4964b1048","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 22:04:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93a9e5b5d73640e59a1d8920b32a8aaa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1772e13ba6aa4162b9c2d9d290c9f1e7 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"951c61f91c399247ef490d96d802a15afd555594","revisions":{"16ee7fc1a20b53e04ba371d6d105e121bd367e28":{"kind":"REWORK","_number":1,"created":"2021-08-12 08:29:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/804362/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/804362/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/804362/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/804362/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/804362/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/804362/1"}}},"commit":{"parents":[{"commit":"d86d74916b7f7619178026c4b8f7825ce099885d","subject":"Add conf file for oslo-config-generator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d86d74916b7f7619178026c4b8f7825ce099885d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-12 08:26:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-12 08:27:05.000000000","tz":540},"subject":"Document available choices using oslo.config api","message":"Document available choices using oslo.config api\n\n... so that the description is properly included in the conf file\ngenerated by oslo-config-generator.\n\nChange-Id: I45fd838fd5877a7da8845c100f54d2d086112c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16ee7fc1a20b53e04ba371d6d105e121bd367e28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16ee7fc1a20b53e04ba371d6d105e121bd367e28"}]},"branch":"refs/heads/master"},"301920b2f5bef03c79b296d4788aa9299f703eb6":{"kind":"REWORK","_number":2,"created":"2021-08-12 08:32:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/804362/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/804362/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/804362/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/804362/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/804362/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/804362/2"}}},"commit":{"parents":[{"commit":"d86d74916b7f7619178026c4b8f7825ce099885d","subject":"Add conf file for oslo-config-generator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d86d74916b7f7619178026c4b8f7825ce099885d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-12 08:26:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-12 08:31:45.000000000","tz":540},"subject":"Document available choices using oslo.config api","message":"Document available choices using oslo.config api\n\n... so that the description is properly included in the conf file\ngenerated by oslo-config-generator.\n\nChange-Id: I45fd838fd5877a7da8845c100f54d2d086112c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/301920b2f5bef03c79b296d4788aa9299f703eb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/301920b2f5bef03c79b296d4788aa9299f703eb6"}]},"branch":"refs/heads/master"},"c880fddc15c830fae9aa46c2c37e3d581fda4d5f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-06-21 23:09:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/804362/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/804362/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/804362/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/804362/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/804362/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/804362/3"}}},"commit":{"parents":[{"commit":"7897a0400abf6767d83361f287d3e636ca113051","subject":"Add missing [cinder] parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7897a0400abf6767d83361f287d3e636ca113051"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-12 08:26:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-21 23:09:19.000000000","tz":0},"subject":"Document available choices using oslo.config api","message":"Document available choices using oslo.config api\n\n... so that the description is properly included in the conf file\ngenerated by oslo-config-generator.\n\nChange-Id: I45fd838fd5877a7da8845c100f54d2d086112c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c880fddc15c830fae9aa46c2c37e3d581fda4d5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c880fddc15c830fae9aa46c2c37e3d581fda4d5f"}]},"branch":"refs/heads/master","description":"Rebase"},"951c61f91c399247ef490d96d802a15afd555594":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-06-21 23:40:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/804362/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/804362/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/804362/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/804362/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/804362/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/804362/4"}}},"commit":{"parents":[{"commit":"c174312456b83ca0afc42558a13aacad82a2a353","subject":"Add missing [cinder] parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c174312456b83ca0afc42558a13aacad82a2a353"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-12 08:26:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-21 23:39:45.000000000","tz":540},"subject":"Document available choices using oslo.config api","message":"Document available choices using oslo.config api\n\n... so that the description is properly included in the conf file\ngenerated by oslo-config-generator.\n\nChange-Id: I45fd838fd5877a7da8845c100f54d2d086112c71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/951c61f91c399247ef490d96d802a15afd555594"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/951c61f91c399247ef490d96d802a15afd555594"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
