)]}'
{"id":"openstack%2Foctavia~805955","triplet_id":"openstack%2Foctavia~master~Ia8ec48ab858eeecf71ed429e5b7625fd7af9d8f6","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-02-08 10:04:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia8ec48ab858eeecf71ed429e5b7625fd7af9d8f6","subject":"Preserve haproxy server states during reloads","status":"MERGED","created":"2021-08-25 07:32:57.000000000","updated":"2022-02-08 10:06:28.000000000","submitted":"2022-02-08 10:04:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805955","meta_rev_id":"73cf0e45e143491f7ee81fbd36bcf522888aba19","_number":805955,"virtual_id_number":805955,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-08 10:04:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-02-08 07:25:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-10-26 07:25:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2022-01-26 23:49:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2022-02-08 07:24:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-02-08 07:24:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-01-28 07:00:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-25 10:16:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-06 21:09:24.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-10-26 07:25:51.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2022-02-08 07:24:48.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"20442f8a4f29002efcc06ff1d6af86439e38512b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-08-25 07:32:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cbc0ae10ea7f227c7e28a9064149b07d27a5a252","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-08-25 07:36:55.000000000","message":"Patch Set 1: Workflow-1\n\nWIP, missing release note...\n\nPlease note that the load-server-state-from-file behavior was \"slow\" before haproxy 2.1 (people complained that it was taking 1 hour with 100k backends), it was improved in https://github.com/haproxy/haproxy/commit/da29fe2360e61a5ed4acd283765b20addd5a3ea8\nUsing it with haproxy \u003c2.1 should be safe with a low number of backends and servers.","accounts_in_message":[],"_revision_number":1},{"id":"76c12170ad6cdb895262bff4a954e0bf5ed3c760","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 10:16:03.000000000","message":"Patch Set 1: Verified-1\n\n(11 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d2e0f81ab9d44f3b197b2aab09c2b27 : FAILURE in 7m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebb04b60139e4d8f93f7d8f19e35f246 : FAILURE in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d68794bc92954daabc9aa530529e544f : FAILURE in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02b0ce8a95f94274b1714e9c0e95e1ec : FAILURE in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50329250877447cb9f986a7a705ecb15 : FAILURE in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e066f747f90410593cb80f6c6901b7e : FAILURE in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e11f909b4764451953f00d6697cdfe6 : SUCCESS in 13m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/60b54e606c024bb691ba53131e4cc65d : FAILURE in 7m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5d5aebfce7204316a94d533de33bfd88 : SUCCESS in 10m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/da8a9a9fdfa940b2b7766320fc1057d8 : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9334ef9f42d24753b6807090ca7799c6 : SUCCESS in 10m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/be9f0f58b4a54c858c733e5433e62dd2 : SUCCESS in 2h 07m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3b5358ae8a684fd6887fd1c3404e6ebc : FAILURE in 2h 09m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2fba7be26c2c43f18773604dad879e07 : SUCCESS in 42m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca3c8f69e25c49558d2e4b7d0652fac9 : SUCCESS in 41m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7e0baf5a6ec7477ebbca4afe89ab9e8b : SUCCESS in 1h 07m 05s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6414cc5981764a66a01e3aac3fce7db1 : FAILURE in 2h 04m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0ccd15ad0b674f7f93b50a852a9077ce : FAILURE in 2h 40m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f7d0f4b0de6c48d087bb763938299c73 : FAILURE in 2h 24m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/eea0fd53a4d7428db7eeadd13c38edff : FAILURE in 2h 11m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/21a27c632fc94bdaac6560af2afd9ed9 : FAILURE in 1h 56m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e32b74fc1ef845508c22cc11174ce880 : FAILURE in 13m 05s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/certificates/manager/local.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/compute_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/compute_tasks.py\n  Comments left for invalid file octavia/cmd/health_checker.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/amphora_info.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalived.py\n  Comments left for invalid file octavia/amphorae/backends/health_daemon/health_daemon.py\n  Comments left for invalid file octavia/amphorae/backends/utils/network_namespace.py\n  Comments left for invalid file octavia/amphorae/backends/utils/haproxy_query.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalivedlvs.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":1},{"id":"e667707b496e0ac1037ed2b7fed56b48d972f0ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-08-30 15:01:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"598ad382467e97211c329aee0c2a72d65e466eb0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-08-30 15:14:59.000000000","message":"Patch Set 2: Workflow-1\n\nNeed to evaluate the impact of this feature on haproxy start time.","accounts_in_message":[],"_revision_number":2},{"id":"23de4ccbcd2196407c6db56ccd1733bc257a8ba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 17:14:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e23142ff23247568b7f6eb4741e75c7 : SUCCESS in 10m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55ccf7b9105d4f128a7086d3d4f931e4 : SUCCESS in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8c1ddad303341de96c1e1c1665b706c : SUCCESS in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d79e3ee918a4c04b351d1cf3cd96257 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcf8f3d6aa41473a9a92105da3c48aa1 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b130c8e0f96943f1829aa991922c8769 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/692557ecaf6942909fbee94e2fa588ea : SUCCESS in 14m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f05e8f7d4bb4178a239b7e54ebfa4db : SUCCESS in 8m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f37753d03e294f3cb45a59e3b2b08ae0 : SUCCESS in 8m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2ea5a1f0c5884d15ac8cd0e10e32cc65 : SUCCESS in 10m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/acab8e4205bb4540bb8984673a3c32cf : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dd5dbe1280d2409aa06ef83c97c3c438 : SUCCESS in 9m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/89a103444fc349a0b450150a844ce491 : SUCCESS in 2h 10m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f34bf199e3145d3935e29367898aa5b : FAILURE in 1h 43m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/74c8437da7a44642b90d72098177c8a2 : SUCCESS in 36m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d614e97423342c7a680622a95449503 : FAILURE in 34m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/364b808284794803a945d384921bb7ae : SUCCESS in 50m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/71e7a7e47a8c46f198a1ff878d00fc14 : FAILURE in 1h 54m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/18facb2eb3d54444b5825b86389df216 : FAILURE in 2h 05m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d62664ff8ddc458f9b8fc552d5227b4d : FAILURE in 1h 45m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fcb8a71acae74853973286623e3e8bb2 : FAILURE in 1h 46m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8047aab6bd3047fab7f47ee6726035db : FAILURE in 1h 46m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5b125fdc2d2d4f65a797b28b0325a8db : FAILURE in 10m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bad88116c35162c79c543414007c1f20e824c728","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-06 08:06:18.000000000","message":"Patch Set 2: -Workflow\n\nRemoving WIP, adding this feature doesn\u0027t seem to impact haproxy start time (tested with a pool with 1k members)","accounts_in_message":[],"_revision_number":2},{"id":"92b7f39bc79278ec2526aac1447de760f9464305","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-06 12:58:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"401522affa66daf7d4f8f320e541fe29659fb0bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 15:42:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10434f17669c420aa1e4f879be24b3f4 : SUCCESS in 13m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f4d5c986c5b42f098e7a627372abb29 : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef5ad31c830b49d49c5aa90945808b85 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/510a1f12180c452ebeb3f935d70f90cc : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/809ad92806dc4188a565c1bf6d0ce973 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6200354f623c45f481532f8139ecd8f7 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebc7f9b8e7fa49df9927d6621fe00a48 : SUCCESS in 14m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c761cc397d814a8bb5b37805f5553b96 : SUCCESS in 6m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b6ab2588df704988bf29b7358e007bf1 : SUCCESS in 7m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/31e9dae61b8e4e5fa2ade1aa9c880402 : SUCCESS in 10m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/45c04a76500c4f9da150bd3ed9937e4a : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e7bb7075ece94b139d2e340f4d8ff04a : SUCCESS in 9m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/217c0cdb5d634ec1852bc0337aabba81 : SUCCESS in 2h 33m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/efaa9ba449f94f9eb697c2508af605e4 : FAILURE in 2h 14m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d9d0a9cd8c3b477ebb06f5ddcdb38797 : SUCCESS in 38m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9b26b83b3e964b30af860c7903445a8e : SUCCESS in 54m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b53beb03fdf4480a7e4cd0ab85337ef : SUCCESS in 37m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3a2cc6f4feae45e28d5b12e23f6246aa : FAILURE in 2h 04m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/991903629aad47d99db8f6bef2c8202a : FAILURE in 2h 14m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/887a7c3395734244b619e78ea5649649 : FAILURE in 1h 41m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8725efa6bd604e6381673f92cf78d42a : FAILURE in 1h 56m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4f3ccb1f3a7b42c2b0c6a2d446d863a1 : FAILURE in 1h 33m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3a3bc104c080472296f5b5f7676724d0 : FAILURE in 12m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"67f6956b821aa818b36ac1489932ae12095a8a0d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-06 19:17:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"da5c3e526991d92d91aeb0d68f836db1f1f33684","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-06 21:09:24.000000000","message":"Patch Set 2: Code-Review+1\n\nThe code looks fine to me, but I\u0027m a bit concerned by the job failures. \"did not begin passing traffic within the timeout period. Failing test.\"\nI\u0027m wondering if there isn\u0027t a timing/status issue here.","accounts_in_message":[],"_revision_number":2},{"id":"f26aa0257a076f078490fc2e7d49494f12b37069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 22:09:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b3541c8461f4e06b60557cec77ba737 : SUCCESS in 14m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1500b47431c047a1832e81794185201a : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b5c1a55f5dc415cbdb4a1766e70fb65 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f19ca2064d74cb1a8d3405ae962a65b : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67efe249855d4cdbb04a7547ea1d1bf7 : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46a9e87d18f74c74abf2f29a9e011153 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c318fbed9c1a4359ba2c95bfa8c1fc5c : SUCCESS in 14m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22196142a1eb4ee6adfb0c4dda8eecf5 : SUCCESS in 5m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b46f072f459d4a8c96e48c1e67a29202 : SUCCESS in 7m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0354f844edbf4ee39886fe7f84c83cc3 : SUCCESS in 10m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7df7d7dced564163b54a27bd25b36ba5 : SUCCESS in 5m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b9a5f305957743a5bebe13474442da0c : SUCCESS in 8m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2ab7865e192f4557aa486a1edb8a93d8 : SUCCESS in 2h 44m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/69df8f7d47bd4e44b10ed3804ec4e921 : FAILURE in 1h 45m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/187864d804cb499babaaeb49136b0021 : SUCCESS in 35m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2490162ec9ea41579753a96c841fb1b8 : SUCCESS in 51m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9de5fdcb0ddf4bf28c84160c29aeef7f : SUCCESS in 37m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8fb07bee43584ed99eb5cd9e78d8a237 : FAILURE in 1h 34m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/43fa39428cf6417fa466e1e5598ec7f3 : FAILURE in 1h 58m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/14eb88cf66a145baa07c1669d9d2dcf3 : FAILURE in 1h 48m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1fdc9efe0f9f4293aa20a525570161f9 : FAILURE in 1h 49m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9710743f26c84fb7902b3a4fddaeb500 : FAILURE in 1h 36m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5519acaad8b74df0b6cd726d978efc02 : FAILURE in 8m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f870693546c186ba6152460976efafa538fb6e75","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-07 06:18:00.000000000","message":"Patch Set 2: Workflow-1\n\n\u003e Patch Set 2: Code-Review+1\n\u003e \n\u003e The code looks fine to me, but I\u0027m a bit concerned by the job failures. \"did not begin passing traffic within the timeout period. Failing test.\"\n\u003e I\u0027m wondering if there isn\u0027t a timing/status issue here.\n\nYes, there\u0027s an issue with my commit, after deleting a health monitor, the members that were DOWN are still considered as DOWN. I\u0027m trying to find a way to reset this state.","accounts_in_message":[],"_revision_number":2},{"id":"a47e177177525d323d116eb51e493ae6f5a5c4fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-08 13:02:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d2a41031790b76e4f562102576654b677d085531","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-08 13:05:42.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2: Workflow-1\n\u003e \n\u003e \u003e Patch Set 2: Code-Review+1\n\u003e \u003e \n\u003e \u003e The code looks fine to me, but I\u0027m a bit concerned by the job failures. \"did not begin passing traffic within the timeout period. Failing test.\"\n\u003e \u003e I\u0027m wondering if there isn\u0027t a timing/status issue here.\n\u003e \n\u003e Yes, there\u0027s an issue with my commit, after deleting a health monitor, the members that were DOWN are still considered as DOWN. I\u0027m trying to find a way to reset this state.\n\nI fixed this issue by moving the load-server-state-from-file directive from the default section to the backend section, and now it is present only if a health-monitor exists.\n\nIt means that if the user deletes a health-monitor, the state file is not reloaded. The members are considered as UP even if they were down before deleting the HM, this is the behavior we expect.","accounts_in_message":[],"_revision_number":3},{"id":"5da0a0cb5cc8d8c43e8746075477966fe68b448b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 15:57:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43cfca72d84b41b4a11102b37d8ab517 : SUCCESS in 18m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5853d76da62f42819e8de8d85da9e8b2 : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc5b1c54aece4f72b7f0b61f313f9983 : SUCCESS in 10m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/017b35aa8dfd494d9c48739fb416fd5c : SUCCESS in 13m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e64340c9e87b437db7f96b048a1357be : SUCCESS in 13m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12ee2fbe81c5472fa5d6a8f426fcd538 : SUCCESS in 10m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ff16c7f53fc4c14ab0e80428904bab5 : SUCCESS in 13m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b142e49c93f4fa4869b2742dc1c37fa : SUCCESS in 7m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ef7c5d662c8247bcbbbeba99f8833720 : SUCCESS in 13m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1190abb0737443d4b24173e43ace54ab : SUCCESS in 15m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c915a55b06974d03888e37dc7515f1a4 : SUCCESS in 7m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/86b7075192ff44e2bea28db57a709480 : SUCCESS in 17m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5842f6472eaa4f83a17dfde2687b4b40 : SUCCESS in 2h 16m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3fe2d6eebd10430190ba57e0f38daa3b : SUCCESS in 1h 54m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f3aefe0e9a8b46eeb098eb7417750137 : SUCCESS in 48m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/046f08db9afc47aeb4b1dd87de9604c4 : SUCCESS in 1h 10m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f9b53ecaa02f47d391d2354e0ba398f8 : SUCCESS in 42m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/09eb31b0f7fb42849a8d81ae7e25acf8 : SUCCESS in 1h 56m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d0d0d0f65d8f430d81265ec16df32f97 : SUCCESS in 2h 49m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d44aedfdde40476c89aa03f257e171f4 : SUCCESS in 2h 32m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5294d557d4484ed79d4aea388f92aedf : SUCCESS in 2h 25m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/eccab6e8fc664493abc9f759c8f94df5 : SUCCESS in 1h 59m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a451c477eef146f8b0cbdca97a194cc0 : FAILURE in 12m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"de03947accfdd8d4c468d8f2fe9acb8067e22ca1","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2021-10-26 07:25:51.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"704ae1130c35b7a3ee2a693843454dcb74a0da99","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-26 23:49:32.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bea08ef8ba912ac29a6ab81183524d684dbb069b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 02:17:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83fa99cebe1d4bfd880998ba1ee4de6f : SUCCESS in 9m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55cdf54e1aaf46038a2d9f8da57e0819 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3adc0bdedba54800aebe319e69da5874 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10447e20a233475e8d84850b4c943432 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f136f899d60441de959dc55a8b4d46ca : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78a3af0b35524d79b5409fe531807c54 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6d2cb7efca44bfcb9b7626f35837688 : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56f03f8ed9e34435964d442428b23aa5 : SUCCESS in 5m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5dc78f28ca724774aa2c3fa78b27d4f9 : SUCCESS in 7m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ceb15c9a1d844b88b0f1a996b185cf63 : SUCCESS in 9m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1da7f23ef68045558a28de16234d543e : SUCCESS in 5m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef9af565e6df40d689a6ded63307e4b3 : SUCCESS in 7m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9b4627b26a844abaa6960e3679990a29 : SUCCESS in 2h 19m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0747a0dfc8a74945af170d0aed158a1b : SUCCESS in 2h 18m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/424a6785f16b4a0b99e14b85ff6ea837 : SUCCESS in 44m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/98cab25070d8438b89cb9043924c71c3 : SUCCESS in 1h 14m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c851ea47efe64384af5504fe8b895f5b : SUCCESS in 43m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5c561417652b45c5824f64ded72809e1 : SUCCESS in 2h 15m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3259777ba84e48b092907b4a3ace9bf7 : FAILURE in 2h 23m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5f670431198d447684ee23abca68592d : SUCCESS in 1h 41m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/47fb6c8695b14c9183733ceea2b74391 : SUCCESS in 2h 00m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/20dfe44950d746e7b150d7b3a11f3b1a : FAILURE in 1h 54m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/a997af2feb8248649e12d71df647eab4 : SUCCESS in 2h 20m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3e7a15d6a73351090750aca13b9444f4a773d932","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-28 07:00:14.000000000","message":"Patch Set 3: Backport-Candidate+2","accounts_in_message":[],"_revision_number":3},{"id":"da9715e5c7775be2ffdd320cebb349c58695ab82","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-02-08 07:24:48.000000000","message":"Patch Set 3: Backport-Candidate+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f52e70e03391bd9c654f355327a0048b48a66ed6","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2022-02-08 07:25:01.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d53cf1afa919bddfc9413fb0f860a63445c3271c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 07:30:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"09aceb97af721976593544c62ba2a918130fdb2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 10:04:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6ac6600b67e451787bddc91898d5516 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f2f93eb07914901a46139dd70c93986 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5fc8e491a714a76a6c56059f13976b1 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed49d83a23134a5bb435af9a48df9f7a : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8660d1df93041c89dbfe6607ab3842e : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d09e692e72a482dbc64b5fb7dae2ddd : SUCCESS in 5m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5852945863304c058fef2d51935f723c : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/603189879df8447187f4afd126a71cca : SUCCESS in 6m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/858748b30f2d4e30a69862631a54b1c4 : SUCCESS in 2h 29m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec8b1341f32c4bb7be2fdc31f04d0500 : SUCCESS in 1h 24m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/45d2df05e3cb400384d1237a6c108ce7 : SUCCESS in 35m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/075485ec3a5540cf9f73558e5225072f : SUCCESS in 45m 06s","accounts_in_message":[],"_revision_number":3},{"id":"c6634104fa63591fb8220875794a7929808feb52","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 10:04:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"73cf0e45e143491f7ee81fbd36bcf522888aba19","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 10:06:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d123780c09894b1eb69feb65d9e9638f : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b140346a91cd4d5a9d1e310d76e666e9 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1b15e1ef22963324e556e072f4bc58021f96bfec","revisions":{"d1c53f059da66a1915d161b24f466c9731a292f9":{"kind":"REWORK","_number":1,"created":"2021-08-25 07:32:57.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/55/805955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/55/805955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/805955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/805955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/805955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/55/805955/1"}}},"commit":{"parents":[{"commit":"5c8f60614e97b76d95b7f9b90bc142de4dea1cdd","subject":"Merge \"Alias change amphorav2 -\u003e amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c8f60614e97b76d95b7f9b90bc142de4dea1cdd"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-08-23 17:53:28.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-08-25 07:26:51.000000000","tz":120},"subject":"Preserve haproxy server states during reloads","message":"Preserve haproxy server states during reloads\n\nhaproxy doesn\u0027t keep the state of the servers (UP or DOWN) during\nreboots, so a member with an ERROR operating_status may be updated\nto ONLINE after updating a load balancer.\n\nThis commit adds the load-server-state-from-file option in haproxy\nconfiguration files. It also adds an extra step in the haproxy systemd\nservice file to dump the current server-state in a file when reloading\nthe service.\n\nStory: 2009142\nTask: 43084\n\nChange-Id: Ia8ec48ab858eeecf71ed429e5b7625fd7af9d8f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1c53f059da66a1915d161b24f466c9731a292f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1c53f059da66a1915d161b24f466c9731a292f9"}]},"branch":"refs/heads/master"},"aed289df2107ed30574f4a79f6e59704add25589":{"kind":"REWORK","_number":2,"created":"2021-08-30 15:01:45.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/55/805955/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/55/805955/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/805955/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/805955/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/805955/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/55/805955/2"}}},"commit":{"parents":[{"commit":"20da8e1aa3c56e9e25191abde8fae4e39aeff529","subject":"Fix pylint checks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/20da8e1aa3c56e9e25191abde8fae4e39aeff529"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-08-23 17:53:28.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-08-30 14:59:28.000000000","tz":120},"subject":"Preserve haproxy server states during reloads","message":"Preserve haproxy server states during reloads\n\nhaproxy doesn\u0027t keep the state of the servers (UP or DOWN) during\nreloads, so a member with an ERROR operating_status may be updated\nto ONLINE after updating a load balancer.\n\nThis commit adds the load-server-state-from-file option in haproxy\nconfiguration files. It also adds an extra step in the haproxy systemd\nservice file to dump the current server-state in a file when reloading\nthe service.\n\nStory: 2009142\nTask: 43084\n\nChange-Id: Ia8ec48ab858eeecf71ed429e5b7625fd7af9d8f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aed289df2107ed30574f4a79f6e59704add25589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aed289df2107ed30574f4a79f6e59704add25589"}]},"branch":"refs/heads/master"},"1b15e1ef22963324e556e072f4bc58021f96bfec":{"kind":"REWORK","_number":3,"created":"2021-10-08 13:02:56.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/55/805955/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/55/805955/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/805955/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/805955/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/805955/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/55/805955/3"}}},"commit":{"parents":[{"commit":"8ac5aa7cbec259d79b63e8777ee26bd8391c33b5","subject":"Fix docs jobs and pep8 errors with pylint 2.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ac5aa7cbec259d79b63e8777ee26bd8391c33b5"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-08-23 17:53:28.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-08 13:01:19.000000000","tz":120},"subject":"Preserve haproxy server states during reloads","message":"Preserve haproxy server states during reloads\n\nhaproxy doesn\u0027t keep the state of the servers (UP or DOWN) during\nreloads, so a member with an ERROR operating_status may be updated\nto ONLINE after updating a load balancer.\n\nThis commit adds the load-server-state-from-file option in haproxy\nconfiguration files. It also adds an extra step in the haproxy systemd\nservice file to dump the current server-state in a file when reloading\nthe service.\n\nStory: 2009142\nTask: 43084\n\nChange-Id: Ia8ec48ab858eeecf71ed429e5b7625fd7af9d8f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b15e1ef22963324e556e072f4bc58021f96bfec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b15e1ef22963324e556e072f4bc58021f96bfec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
