)]}'
{"id":"openstack%2Foctavia~807119","triplet_id":"openstack%2Foctavia~master~I2d67649fa6d3be87204680ba5778e0519ebfcbcf","project":"openstack/octavia","branch":"master","attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2021-10-26 07:19:29.000000000","reason":"Quentin GROLLEAU replied on the change"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2021-10-20 06:46:05.000000000","reason":"Gregory Thiemonge replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I2d67649fa6d3be87204680ba5778e0519ebfcbcf","subject":"Re-enable active-standby job as voting jobs","status":"NEW","created":"2021-09-02 11:21:22.000000000","updated":"2022-10-26 16:19:02.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cb2a4bc22d7116a79f2a8bb0279f709982992900","_number":807119,"virtual_id_number":807119,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-10-20 09:24:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":1,"date":"2022-10-26 16:19:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2022-10-26 11:35:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-10-21 08:18:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-02 11:21:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-18 09:01:08.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-10-19 14:02:15.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-10-26 07:19:29.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2022-09-29 15:02:56.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2022-10-21 08:18:38.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2022-10-26 11:35:41.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"50a82c91bc33d729195a4040b1ad615a41db3780","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-02 11:21:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8570b1d711145243fa6a8cbccf7df26eb9ad66c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 11:21:46.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/octavia/zuul.d/projects.yaml@master\", line 43, column 31","accounts_in_message":[],"_revision_number":1},{"id":"4304b8d1c80c84f0d12aa41b1e88a507e826b3b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-02 11:36:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2d29b5de437e984d2db8caeb08d72351099e07b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 12:22:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ec924d199a741698210e38ae50c5f93 : FAILURE in 16m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d22d2b55c2bb40e0b691aca2a8b4e7d5 : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d78112d4f4b4c56bcd539054003fad3 : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c45f5971961b4c9a801abe31069022ec : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf7efd5ec5d541c1b2e887cb783d8210 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b8ddc646c864a9f83774e96bd63ea2c : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/752a7bde366b46f38afe626bf89111fc : SUCCESS in 13m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1ff3c3af8b6f403d9a0a29df60417681 : SUCCESS in 7m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a662ff2b039d4377b13693cbde0de037 : SUCCESS in 8m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/17ce2eddb6354a0caf96d0948231a1bf : SUCCESS in 34m 41s","accounts_in_message":[],"_revision_number":2},{"id":"b6189d5c25b53ae7791a7dd1deb932fa4d83a4f8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-02 12:24:18.000000000","message":"Patch Set 2:\n\nrecheck\n\nread timeout in cover","accounts_in_message":[],"_revision_number":2},{"id":"74aea682b561334768112d7f68f483aab7784ef1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 13:05:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/741367ad34624f0ea4c9b2c9ae6bea11 : SUCCESS in 11m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93ad288d88194031b1ef2bdd8c56bd3a : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8c4b72342cf432da0644810a1737743 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0604ecfbfbdc4d6780cdd73738739fba : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39405d68eceb4e88b41cf3541c592474 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2c125d3edbf40c393a8891d335c2392 : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a20fa1c9307e4fe793339f97d5b65b49 : SUCCESS in 15m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bb24571dc3164def9038dabe4ca1c115 : SUCCESS in 10m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c75acb32706d41cf99de841fbbf86332 : SUCCESS in 10m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/043a47c9cc35421c851298c0e4117474 : SUCCESS in 35m 29s","accounts_in_message":[],"_revision_number":2},{"id":"81d7521bb0218c910a0a7056cd8b72ee27b340bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-11 07:26:08.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"dceae1892b9c4cd82b6cd2782386f141031ebf0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-11 08:59:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/036b7b330d494f3b99033330c6a45c4d : SUCCESS in 11m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9abe9b2fc3534f758ea13b468f807cf5 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a76a38b7091e41c8bfabea9fb33ebb37 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a47fe4b9d7844496880802fd027a57cc : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a718263b7e994dd7b64ab1049c854e33 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf839bfbd1b74e9a8e793a33b99dddcd : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f9579975d8a48b386d442b11f58846b : SUCCESS in 12m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e2db09bc641b4d57b2c958bf6ba2e0d3 : SUCCESS in 5m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f7dcafeecf7c489198c5ca7345d5f005 : SUCCESS in 11m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54d0bbd380c8432c97b42376a042c066 : SUCCESS in 35m 19s","accounts_in_message":[],"_revision_number":3},{"id":"2e7e7ff28ecc00ded7a4f37971b33aa1108809c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-15 14:14:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"45883c98db2bea237e86304f8f42bad68a70166a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 15:02:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f06977b255614cedadaebf34cf112a16 : SUCCESS in 18m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/231f0cacdc064025ae92630a31ed5825 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ad2b06f964d4f46ab3280edca214470 : SUCCESS in 12m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81141b287c74404da4dcf804a658e7d5 : SUCCESS in 15m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc7ec1d6ec4d4e98bcd3b4cff1f5414d : SUCCESS in 12m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26d4732d2d6b4841a14b8ac4dc9c292f : SUCCESS in 9m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f938e1c9eab2407d8aa3d21630e43011 : SUCCESS in 12m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ae4ad778c59f4b0eba21758d7e207177 : SUCCESS in 15m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/03d04741d86c4ec58def82e47a87a758 : SUCCESS in 20m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/77948f88573141abaa5dd8f2b6bf9f7a : SUCCESS in 45m 41s","accounts_in_message":[],"_revision_number":4},{"id":"aaa347ea296cc6c8d04cff0691f623a6e8a33ba6","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-10-18 09:01:08.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5479e6e24d7c27c7fb39ff3b175cc0771589ba20","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-19 14:02:15.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"21943d9be88d180ec76d72e664b4e6d6270cbd68","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-19 14:02:50.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"3b536f61e9703dc76c8b143c25c3ef5e6c092793","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-20 06:45:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cf56d093ed9c09f47499ef3774b24cca7f2b6419","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-20 06:46:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"19efb8de56f91e292a2d5f6c7c07ffc1c66d8bea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 09:37:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47f2b672a7894360ad076bed5993bae7 : SUCCESS in 13m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8b2a551a2c54773af399eb8ae39c29a : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd7bbd5634014936a7ccb97299281ee2 : SUCCESS in 8m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bcf9f48ad43476fa4437d28b60b2306 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a3bcea369a74a91b9bf710314417c87 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bd190b1d1484dd88f19092fbcdc0f34 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1f414ea7b0544738e79e0e2b7e74bb8 : SUCCESS in 11m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cd858200b87d4f43842e9fd8a20349e7 : SUCCESS in 7m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/de37187d4240460dad8b18e4b96aa259 : SUCCESS in 9m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f2ec109dd1aa48389852bdf548da9a32 : SUCCESS in 7m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7f3a17e4d6404d05b404af06496a2dc5 : SUCCESS in 9m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/df96944dae5e46f4ae6a44929e2e7dea : SUCCESS in 2h 02m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f69d9bce815a4453a9fd0eb607d625b2 : SUCCESS in 1h 40m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7be0858e1fbd4a17b9e9f48f985618f2 : SUCCESS in 40m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e4a324b1907e4824bd1b61d93bba19d1 : SUCCESS in 35m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/492879c4dfb240e5babd62442bf9a419 : SUCCESS in 52m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/207789cae3d04543970157851dc50929 : SUCCESS in 1h 23m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1e87dc61e84044c59dca6e6b0494badc : SUCCESS in 1h 53m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cb844accc98b483abe6e931eff2eda5f : SUCCESS in 1h 34m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/020a1823e17f4dfd9cd3e2c6cfcdf59b : SUCCESS in 1h 31m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2893a25f0597497f9b2f845140587f53 : SUCCESS in 1h 22m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4da41509ebbb4805ad3e7822af26186a : FAILURE in 13m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d513947a6af7955fa6b8499c9e07b1b9e8f3891a","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2021-10-26 07:19:29.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"2ecd00c74083685f88fe55e4e1a14def69db3838","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-09-29 15:02:56.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e023d1536e57a859214a26cc58a1aa7c190c184c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-10-20 07:19:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a43839a48922e9df5b9ce0a5a12122665903eb2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 09:24:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f841fabcb0dd4117b4bb8577a3b66e19\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a02368d4cf94494cb0cbba5b69040cb3 : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7633fd6a93ed487484927edae62ffd0a : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2de358530b024da3a272af4e90033533 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0106e9863fa4c85ab4960e8c5a02210 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50137b09ad28483dbb6ebaaf81ee2cb8 : SUCCESS in 12m 24s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/eb21e1a4ddb24109a52268487200da2c : SUCCESS in 5m 10s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4fac30b5b3174b41b752bb4e4868ff4f : SUCCESS in 7m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/25bbf711809240d4be60fbea06df0829 : SUCCESS in 4m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c96af670a18a45599236e3a44e8c5dc4 : SUCCESS in 6m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d5c22765764342b1997420d1ca97db81 : SUCCESS in 7m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2e92bbcee60b4ae7806fc1ea401f65e0 : SUCCESS in 7m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5e5519b5dcce411392b98b030e380c8e : SUCCESS in 1h 27m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e66a70292f245a8bd7e8510e2b8e6c2 : SUCCESS in 1h 59m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5c010c7809f3452c804d8d7775dd66c4 : SUCCESS in 36m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e7a8a14f4924dcc8896e99d7c58afd6 : SUCCESS in 37m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/57cd60032a0141c2a1c800ebd8d0d9de : SUCCESS in 1h 03m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/48bb788d39d44f5091b72f46b5cf46d7 : SUCCESS in 1h 09m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/21c5d26e16bd4149a5af4ee278a49449 : SUCCESS in 1h 50m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/48cf61fe0b504cd48661e0585fdbd7f6 : SUCCESS in 1h 54m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/75ca3fbab8534adabdc6394c5b0fdd86 : SUCCESS in 2h 02m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d4e3adca0a52490e81f9c8920747bcfb : SUCCESS in 1h 51m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cd6f7284d7f54006b6c7efd9505434eb : SUCCESS in 1h 42m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7bab619e18ab407aba32fe8df3548ea3 : SUCCESS in 1h 55m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"cd8541d968f3a5b327c02a476f270890e6ba3cb1","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-10-21 08:18:38.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"be739b6412e81ef3c6f6762c75cff148b34b92bd","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-10-26 11:35:41.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"cb2a4bc22d7116a79f2a8bb0279f709982992900","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-10-26 16:19:02.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"53e19ea485eb25350a66d9459cf3b5b8eb9b6a81","revisions":{"0ed9aa14df3539496712650839314a6260d83f29":{"kind":"REWORK","_number":1,"created":"2021-09-02 11:21:22.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/807119/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/807119/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/807119/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/807119/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/807119/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/807119/1"}}},"commit":{"parents":[{"commit":"5dd7ad9ad8998c5285fe60d9ee4210f87ed45ccd","subject":"Add generic network interface management in the amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5dd7ad9ad8998c5285fe60d9ee4210f87ed45ccd"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-02 10:21:33.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-02 11:20:42.000000000","tz":120},"subject":"WIP Re-enable active-standby job as voting jobs","message":"WIP Re-enable active-standby job as voting jobs\n\nThis reverts commit 39735ebf10426a639fd85565b8e7e752b15d77f7.\n\nDepends-On: https://review.opendev.org/c/openstack/placement/+/807014\nChange-Id: I2d67649fa6d3be87204680ba5778e0519ebfcbcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ed9aa14df3539496712650839314a6260d83f29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ed9aa14df3539496712650839314a6260d83f29"}]},"branch":"refs/heads/master"},"e54afa3888b199edf09b6151f79d47579023a145":{"kind":"REWORK","_number":2,"created":"2021-09-02 11:36:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/807119/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/807119/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/807119/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/807119/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/807119/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/807119/2"}}},"commit":{"parents":[{"commit":"5dd7ad9ad8998c5285fe60d9ee4210f87ed45ccd","subject":"Add generic network interface management in the amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5dd7ad9ad8998c5285fe60d9ee4210f87ed45ccd"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-02 10:21:33.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-02 11:36:13.000000000","tz":120},"subject":"WIP Re-enable active-standby job as voting jobs","message":"WIP Re-enable active-standby job as voting jobs\n\nThis reverts commit 39735ebf10426a639fd85565b8e7e752b15d77f7.\n\nDepends-On: https://review.opendev.org/c/openstack/placement/+/807014\nChange-Id: I2d67649fa6d3be87204680ba5778e0519ebfcbcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e54afa3888b199edf09b6151f79d47579023a145"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e54afa3888b199edf09b6151f79d47579023a145"}]},"branch":"refs/heads/master"},"2f84ee1d96f81293423045198ab618d041151fc2":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-09-11 07:26:08.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/807119/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/807119/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/807119/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/807119/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/807119/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/807119/3"}}},"commit":{"parents":[{"commit":"5dd7ad9ad8998c5285fe60d9ee4210f87ed45ccd","subject":"Add generic network interface management in the amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5dd7ad9ad8998c5285fe60d9ee4210f87ed45ccd"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-02 10:21:33.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-11 07:25:39.000000000","tz":0},"subject":"WIP Re-enable active-standby job as voting jobs","message":"WIP Re-enable active-standby job as voting jobs\n\nThis reverts commit 39735ebf10426a639fd85565b8e7e752b15d77f7.\n\nChange-Id: I2d67649fa6d3be87204680ba5778e0519ebfcbcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f84ee1d96f81293423045198ab618d041151fc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f84ee1d96f81293423045198ab618d041151fc2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8a8d9d0e2a894eed961709ba391bc0da9ca15413":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2021-10-15 14:14:30.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/807119/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/807119/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/807119/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/807119/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/807119/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/807119/4"}}},"commit":{"parents":[{"commit":"0b84631948bd3c5eb25a0921d5fbad657036304f","subject":"Merge \"Increase TCP buffer maximum and MTU discovery\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b84631948bd3c5eb25a0921d5fbad657036304f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-02 10:21:33.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-15 14:14:23.000000000","tz":120},"subject":"Re-enable active-standby job as voting jobs","message":"Re-enable active-standby job as voting jobs\n\nThis reverts commit 39735ebf10426a639fd85565b8e7e752b15d77f7.\n\nChange-Id: I2d67649fa6d3be87204680ba5778e0519ebfcbcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8a8d9d0e2a894eed961709ba391bc0da9ca15413"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8a8d9d0e2a894eed961709ba391bc0da9ca15413"}]},"branch":"refs/heads/master"},"b6056a5ac408d5bca499299eeffdde453c0405f0":{"kind":"REWORK","_number":5,"created":"2021-10-20 06:45:12.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/807119/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/807119/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/807119/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/807119/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/807119/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/807119/5"}}},"commit":{"parents":[{"commit":"0b84631948bd3c5eb25a0921d5fbad657036304f","subject":"Merge \"Increase TCP buffer maximum and MTU discovery\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b84631948bd3c5eb25a0921d5fbad657036304f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-02 10:21:33.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-20 06:44:35.000000000","tz":120},"subject":"Re-enable active-standby job as voting jobs","message":"Re-enable active-standby job as voting jobs\n\nThis reverts commit 39735ebf10426a639fd85565b8e7e752b15d77f7.\n\nChange-Id: I2d67649fa6d3be87204680ba5778e0519ebfcbcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b6056a5ac408d5bca499299eeffdde453c0405f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b6056a5ac408d5bca499299eeffdde453c0405f0"}]},"branch":"refs/heads/master"},"53e19ea485eb25350a66d9459cf3b5b8eb9b6a81":{"kind":"REWORK","_number":6,"created":"2022-10-20 07:19:42.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/807119/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/807119/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/807119/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/807119/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/807119/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/807119/6"}}},"commit":{"parents":[{"commit":"344d6d76ce4481f30fe295114551243d61e7eea8","subject":"Fix prometheus-proxy service name in Red Hat-based distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/344d6d76ce4481f30fe295114551243d61e7eea8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-02 10:21:33.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-10-20 07:19:23.000000000","tz":120},"subject":"Re-enable active-standby job as voting jobs","message":"Re-enable active-standby job as voting jobs\n\nThis reverts commit 39735ebf10426a639fd85565b8e7e752b15d77f7.\n\nChange-Id: I2d67649fa6d3be87204680ba5778e0519ebfcbcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/53e19ea485eb25350a66d9459cf3b5b8eb9b6a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/53e19ea485eb25350a66d9459cf3b5b8eb9b6a81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
