)]}'
{"id":"openstack%2Foctavia~807140","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I8f0d3ba1d083c7e37e7e26617826400a538a0890","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"I8f0d3ba1d083c7e37e7e26617826400a538a0890","subject":"Fix using host_routes in VIP subnet with amphorav2","status":"MERGED","created":"2021-09-02 13:19:41.000000000","updated":"2021-10-05 11:21:48.000000000","submitted":"2021-10-05 11:17:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807140","meta_rev_id":"7746c9659a9bcd020e2537542d90609e6932dbcb","_number":807140,"virtual_id_number":807140,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-05 11:17:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-09-30 07:05:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-10-05 09:33:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-10-05 09:33:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-02 15:58:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-30 07:05:38.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"ce07b79c9d2c9c63f01561cc6603761a49878662","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-02 13:19:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c647e4ded503ffacf7a0f1b1a859a971503d133d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 15:58:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c09908a24124662ad303ad590c5548e : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c70e54b9612c4fefbef7f78364859ec0 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c61b589370964cd9ae307d2b73e544cd : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85a0eb2fd0fc47719fdc0d12f218ae76 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0fa69c00897424face041920677f69d : SUCCESS in 13m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75d164973c7d466b99bbc3b65c9027a4 : SUCCESS in 7m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/088663a2960a45998119a27eb7bb625a : SUCCESS in 8m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e1076310f4034da4a2edfe593f347805 : SUCCESS in 9m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8b9417ff490a4d41aac4b9cd767a93ff : SUCCESS in 7m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5c4c49674d3c40fe8981a89b7791cc41 : SUCCESS in 11m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/49d10718d64b465c93fc8ce05b6c2b73 : SUCCESS in 1h 34m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d5b6c77313b440edb156ec121f6aba4b : FAILURE in 1h 44m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a628daef20a248ffafc88b7b0e2bbf73 : SUCCESS in 37m 40s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0211a0ca4f914401803a6b818ece94d8 : SUCCESS in 35m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e1d1e32c47c4417886e1898326b014dc : SUCCESS in 1h 00m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9acbb63559764a2489aac7b7eeceb992 : SUCCESS in 50m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2c6f91cf7afb43599e62c02de0454d64 : SUCCESS in 1h 57m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/23ada84892e54d699521a4632d190820 : FAILURE in 36m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/987bc5711f3247a194cc5c6869096c80 : TIMED_OUT in 2h 35m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/74ac8f1bb1054888a035c65ac285feb2 : FAILURE in 1h 25m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fd2d0080652c4a7d867d1b10855eab7a : FAILURE in 1h 36m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/68d0ca26b9664a32a2a3d0b6df375718 : FAILURE in 28m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9a81bb04fe09db471ed546b608034ed44d96e72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-09-30 07:05:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"dd2c178446c3de24c38e7f235f0a17dbbf50c06a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-09-30 07:05:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d839f99af2d086120d7270cf12d1705cae3f548d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 08:54:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/379916bd64e44dc381b29f867e9ad8d5 : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2747139ce7d74d128963d70d8f46830c : SUCCESS in 8m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0f23b3b90ab497596fd8e68d743db4b : SUCCESS in 8m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f901bcfa60494bfcba20be1ac8a29516 : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c48242f2916e4312869bbc0128efd8bd : SUCCESS in 14m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf14dacfee71432fa7471f30d08cb703 : SUCCESS in 8m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/612afd015cd64dd9ac3de5665b07e4cc : SUCCESS in 9m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/204988bce66442fe87c05ed0be9dda83 : SUCCESS in 9m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/faefdce1cf74480697e15b3be5cd7255 : SUCCESS in 8m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/53492db6e81743e99c8671b57990cbde : SUCCESS in 10m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ea3217dcbeab42039eee3a4a87ac57ad : SUCCESS in 1h 42m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7437d5b5e04244c890a92c61faca88b5 : SUCCESS in 1h 24m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cab26022d7f1400d92cb3a7ada0d3025 : SUCCESS in 46m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2e1efe3e7595472fa97129b0faed1d72 : SUCCESS in 32m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/912ca7cbe15847028933e1243bd4c4eb : SUCCESS in 34m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a0dab5f45f6c4a778c0f8f344a522ef6 : SUCCESS in 50m 39s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0961f56b5c1344c5af7007e62494c169 : SUCCESS in 1h 36m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5ec20e0f865f463cbf0f8813c98351ac : FAILURE in 34m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2ac66c184f034d748a3598f13f8ab3cd : SUCCESS in 1h 32m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/93bb53a24aa6469187f1ca50a1991680 : FAILURE in 1h 43m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e775efabb0104b1c8334148d7deb01a6 : FAILURE in 1h 30m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/77c4699d5a1f4ed79f13b243b522798a : FAILURE in 30m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"34bc1d5b8e5c41e9ffec12ab5bf2ce4d236b192e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-05 09:33:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6b38f13af3e0319b4f99b4b97728a59c5f7a333d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 09:34:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5bf43c77618487c1cb1d885ec6275ee99855c2ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 11:17:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b596796b59e2447d9b241316d621cb7f : SUCCESS in 9m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98b7791c34454c138032152f14a06956 : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08523c8710454d7ab86f69baa5bf01ce : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71835d0849464c9ca064d5f0f1f207db : SUCCESS in 13m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d35491ec8e8c42d8af1035d9c2f5877e : SUCCESS in 6m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a48d8604ac1746c9ac6535810a36e259 : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3c2fdb0c664f488c8eb24633d41d68b8 : SUCCESS in 7m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/60132d0cab114855914f8d32d2c2fb8f : SUCCESS in 1h 43m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46454616729444f0bd6c6f752058aa2c : SUCCESS in 1h 27m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/10c40279ed684347872d1e187c1eac5a : SUCCESS in 33m 20s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8f49c3f4501743e987590d554d26d30a : SUCCESS in 43m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8240702366c4ce88d7a44fe3cc9c254 : SUCCESS in 31m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a7aa579cb1a5460792a3ffe7d33a0e47 : SUCCESS in 1h 06m 53s","accounts_in_message":[],"_revision_number":2},{"id":"f3b3409a3d8fff917ae93aae516a964d20bb0f86","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 11:17:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7746c9659a9bcd020e2537542d90609e6932dbcb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 11:21:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/166bbcde27844e308d750d9422d26fee : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01a80fa7ac314f97911177cafdfd812b : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"061ca2879acdbdf6278c004f94e1248411207ee6","revisions":{"7e9818eda8f050eb93ca60417fe069383b15d15c":{"kind":"REWORK","_number":1,"created":"2021-09-02 13:19:41.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/40/807140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/40/807140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/807140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/807140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/807140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/40/807140/1"}}},"commit":{"parents":[{"commit":"f3b3e35af34667e2a7d718cc203f2ee03105c29e","subject":"Merge \"Fix race conditions between API and worker DB calls\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f3b3e35af34667e2a7d718cc203f2ee03105c29e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-06 06:00:36.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-02 13:19:25.000000000","tz":120},"subject":"Fix using host_routes in VIP subnet with amphorav2","message":"Fix using host_routes in VIP subnet with amphorav2\n\nEnsure that amphorae_network_config parameter is not modified in\nAmphoraPostVIPPlug. Modifying it would break amphorav2 persistence by\nintroducing non-serializable elements in the subnet dict\n\nStory: 2008955\nTask: 42580\n\nChange-Id: I8f0d3ba1d083c7e37e7e26617826400a538a0890\n(cherry picked from commit 76606b4c4dc9595a015afb4c997f5d02ac7a7343)\n(cherry picked from commit 6367a849e9c51dc886bcdc7f99f9ee3f0f7a2516)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e9818eda8f050eb93ca60417fe069383b15d15c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e9818eda8f050eb93ca60417fe069383b15d15c"}]},"branch":"refs/heads/stable/victoria"},"061ca2879acdbdf6278c004f94e1248411207ee6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-09-30 07:05:25.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/40/807140/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/40/807140/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/807140/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/807140/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/807140/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/40/807140/2"}}},"commit":{"parents":[{"commit":"2def0e7aa507875ec03f3a716103832f02f6bd73","subject":"Merge \"Fix duplicate SG creation for listener peer port\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2def0e7aa507875ec03f3a716103832f02f6bd73"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-06 06:00:36.000000000","tz":120},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2021-09-30 07:05:25.000000000","tz":0},"subject":"Fix using host_routes in VIP subnet with amphorav2","message":"Fix using host_routes in VIP subnet with amphorav2\n\nEnsure that amphorae_network_config parameter is not modified in\nAmphoraPostVIPPlug. Modifying it would break amphorav2 persistence by\nintroducing non-serializable elements in the subnet dict\n\nStory: 2008955\nTask: 42580\n\nChange-Id: I8f0d3ba1d083c7e37e7e26617826400a538a0890\n(cherry picked from commit 76606b4c4dc9595a015afb4c997f5d02ac7a7343)\n(cherry picked from commit 6367a849e9c51dc886bcdc7f99f9ee3f0f7a2516)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/061ca2879acdbdf6278c004f94e1248411207ee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/061ca2879acdbdf6278c004f94e1248411207ee6"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
