)]}'
{"id":"openstack%2Foctavia~807144","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I6251cd17bd6fa9faf5109e50c2190dda3614908d","project":"openstack/octavia","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2021-12-02 02:02:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6251cd17bd6fa9faf5109e50c2190dda3614908d","subject":"Update nr_open limit value in the amphora","status":"MERGED","created":"2021-09-02 13:23:23.000000000","updated":"2021-12-02 02:09:46.000000000","submitted":"2021-12-02 02:02:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807144","meta_rev_id":"3ec9dd36f002880af0255a79a59238def3fcaf24","_number":807144,"virtual_id_number":807144,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-02 02:02:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-09-13 07:57:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-10-27 12:51:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2021-12-01 22:01:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2021-12-01 22:01:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-02 15:58:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-13 07:57:42.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-10-27 12:51:53.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2021-12-01 22:01:41.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"7001edbbbc7ae36a52bc8afe22fafe81d91f7bcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-02 13:23:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d91ff4e226dae94a11d43d0c5f92fb37b3fb1042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 15:58:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79c289dd110c4a08b45247efa837a93f : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7074698bd5a549f7844f0a9b2da8e0c0 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eae7dcbd2bd9437bb1f4c4fae07da5f0 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccd3d5dc9c6641cea396d4143dbc13b4 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce56c6dfd25b4a6e830c97692bc66427 : SUCCESS in 14m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/682374c6821b4ae0abd9b2113489ed5b : SUCCESS in 5m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b69766ce984f40d487e528d04b3e595f : SUCCESS in 7m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/356d8ce128c44186af6ee93e0ff4771a : SUCCESS in 10m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/be961f51a21047d896215272e4bd64d8 : SUCCESS in 5m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f7f3cd771ffd49f98ec92dba872f66d6 : SUCCESS in 8m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/34627ad89a6045e5aa242cd63115896f : SUCCESS in 1h 49m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce6bb49250404ef4995d8579a713f79a : SUCCESS in 2h 01m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/14a0913f6dbd447282e753cef6c1a43b : SUCCESS in 36m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a07559b4360e4c3f89324778da61954a : SUCCESS in 41m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8f02485c68748fa9c13b8178d437d3d : SUCCESS in 46m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e0c551e2a600427099803023eb5bfa86 : SUCCESS in 1h 13m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/71faa4a4ad294d54a134613d68d3ddc1 : SUCCESS in 1h 29m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a278040f466444d2a96f9ff256b73d7d : FAILURE in 32m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b55c4d4f04f943899e9eab60d9835958 : SUCCESS in 2h 19m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9cef7649c2fe435d86e74faffc910c65 : FAILURE in 2h 31m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/15305ffe8a594d6ab90bc5cf0f297925 : FAILURE in 1h 34m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8a139e41ca994ad0ad49378884b7409d : FAILURE in 32m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e7b2a1fa0ff8d8746cb439c5a9f00f32b90bb36c","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-09-13 07:57:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4af0d63aa3a95b9feae8c6ccb72fa4469632718f","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2021-10-27 12:51:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7f8c740339557d7e29a9ae408fc8012d3ca0df80","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-12-01 22:01:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6c3233f3ea7f2939607b2c2ce8651f5ed3e9fe1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 22:02:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4a35f5f90664ef19148895cace919d3e7b03652","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 23:49:29.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"392ff66bc625730d64ac7a8107a036a64a9956e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 02:02:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce67d8b193bd4951a29a7994315bbf26 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17e426eabbac4ae6af863915647d45a1 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8237e669f2f949c8a9ba6aca2ac44b7b : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/943b910d926d4df3b72fef91969706f4 : SUCCESS in 13m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/710f17ca92ae435f97b59c70c8572fbf : SUCCESS in 6m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f3c24b95b915418183113cb2de82bec9 : SUCCESS in 5m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cf7f63c066004b7a8b9b931794a058e8 : SUCCESS in 8m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9911f26821314204812136e2c1e776e0 : SUCCESS in 1h 32m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/caf2b61b417d4bebb8cc32fa5d7e0bc6 : SUCCESS in 1h 42m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/09214a0989e0404d8b6709e978788037 : SUCCESS in 35m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d97074346c31409fb4afc2300494986c : SUCCESS in 32m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f5b3023804b5436186c5205d6f458fd0 : SUCCESS in 33m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c672e58ba8684bf4b1711ae3aa706fa9 : SUCCESS in 48m 33s","accounts_in_message":[],"_revision_number":1},{"id":"96e3c137796096bfdaa27e259f0b7094ef72d54f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 02:02:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3ec9dd36f002880af0255a79a59238def3fcaf24","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 02:09:46.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/810ce439bbd341a6a9fcbfeb137b510a : SUCCESS in 1m 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bc4193079ca44fbb2618ea4efbce230 : FAILURE in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c177987f81171845324a71ad6b7bffe6b151379e","revisions":{"c177987f81171845324a71ad6b7bffe6b151379e":{"kind":"REWORK","_number":1,"created":"2021-09-02 13:23:23.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/44/807144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/807144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/807144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/807144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/807144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/807144/1"}}},"commit":{"parents":[{"commit":"f3b3e35af34667e2a7d718cc203f2ee03105c29e","subject":"Merge \"Fix race conditions between API and worker DB calls\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f3b3e35af34667e2a7d718cc203f2ee03105c29e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-22 08:58:08.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-02 13:23:11.000000000","tz":120},"subject":"Update nr_open limit value in the amphora","message":"Update nr_open limit value in the amphora\n\nUsing haproxy 2.x, the ideal rlimit value for nr_open is close to\nconnection_limit * 2.5 (see compute_ideal_* in src/haproxy.c).\nSet this limit to 2,600,000 in the amphora to support a loadbalancer\nwith maxconn 1M.\n\nThis prevents the following warning messages when launching/reloading\nhaproxy:\n\n* \"Cannot raise FD limit to 2375058, limit is 2097152.\"\n* \"FD limit (2097152) too low for maxconn\u003d950000/maxsock\u003d2375058. Please\n  raise \u0027ulimit-n\u0027 to 2375058 or more to avoid any trouble.\"\n\nChange-Id: I6251cd17bd6fa9faf5109e50c2190dda3614908d\n(cherry picked from commit 4174f4a5a47a208d7cb7fa6d736f5bb20e995b64)\n(cherry picked from commit f55376bf5d715e5c1f0b3ace6c2190f8e9def28a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c177987f81171845324a71ad6b7bffe6b151379e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c177987f81171845324a71ad6b7bffe6b151379e"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
