)]}'
{"id":"openstack%2Foctavia~808060","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I4bec17a484d0d39c9cca389421c63d378c36353c","project":"openstack/octavia","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2021-12-02 02:02:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4bec17a484d0d39c9cca389421c63d378c36353c","subject":"Fix PlugVIPAmphora revert function in amphorav2","status":"MERGED","created":"2021-09-09 12:53:06.000000000","updated":"2021-12-02 02:50:13.000000000","submitted":"2021-12-02 02:02:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808060","meta_rev_id":"d9d7cc3bd18f600139e43abc11fa07c5d89a0e95","_number":808060,"virtual_id_number":808060,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-02 02:02:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-09-13 07:35:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-10-27 12:53:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2021-12-01 22:02:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-24 09:48:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2021-12-01 22:02:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-09 15:27:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-13 07:35:13.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-09-24 09:48:40.000000000","updated_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"reviewer":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"state":"REVIEWER"},{"updated":"2021-10-27 12:53:55.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2021-12-01 22:02:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"a6b2e1e4a433f4f72b81fde9f52e3468b42e77e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-09 12:53:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0204d046573c3fd469b28253a249bc1ee0cfcc34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 15:27:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d55ef6cc0e74ec581d30f17fca2397c : SUCCESS in 12m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/282f8e71a6ab4e51aa5706112bfc7d68 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bc887dcb43a428c963a463e5a0e0538 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f10473d87e6745e986f0f9a0d1b282c3 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dcc789583e444578d0094e1d451f2ad : SUCCESS in 12m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4a9d984d6d3e45cea43c695c3aefa31e : SUCCESS in 6m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/92fff7f15dc74c4db4f24397f4802855 : SUCCESS in 10m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/91dafff80e0b4ec4b893dbc92ba85570 : SUCCESS in 6m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/61517a56ffbd402cab1459fce0431ee5 : SUCCESS in 8m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/947396636dc3436bbc56dd37654613d4 : SUCCESS in 1h 36m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c774612802324bbc8c266f9d10a7b555 : SUCCESS in 1h 31m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d11747cc3c704b8d9fa4283385a88b57 : SUCCESS in 37m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0a00f44d79f846f083fd940a92ee215b : SUCCESS in 38m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4a01af94a464e8fba2214bc1c1d8cc6 : SUCCESS in 38m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6dd7aefa1511414aa51b229affe044fb : SUCCESS in 1h 00m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8113afe288c64d14b0b1751cc4b34865 : SUCCESS in 2h 03m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4405791af262474596aff71f226e2b10 : FAILURE in 34m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b45f7c3af75a4c2f895e49f4b2aaedbd : SUCCESS in 2h 13m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/19190f4a10224f7393adc3e223d461eb : FAILURE in 2h 13m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b0ca8a8b3e2a4ede9b86a85df432ff4f : FAILURE in 1h 32m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/185ccf176b86439f8cf833e214beb010 : FAILURE in 31m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd5787bae2542c284e4e0a3140faad693705a5df","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-09-13 07:35:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"86fdac928e1bf7630d30d39b2abe4426025fd077","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-24 09:48:40.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"2e0284b9f652dec574deaff19305bcf798578f81","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2021-10-27 12:53:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dbe00665c7105c5fe398a15630ac0cb7750e339f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-12-01 22:02:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ddd14de5413affa7aef93f49308b84281b89dd55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 22:03:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e569d6dc52d0190baa25b97e804c2427d8b4acb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 23:50:13.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29d7a7d516b3239ce3726001fa1df4e3792ad45b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 02:02:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84c34f8765e84dee87235d2d3d8e1587 : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71299c8c7ecb41299c351c22ad2c7758 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b8c65181a1640bca19a3e6f992f942b : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a86dddfaf38f493c83f71e6bfe998ec5 : SUCCESS in 12m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/543fc04fe0ac46cbb31228a7e29f92e9 : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/95ef6539cf4e4bc1931aec67a506078d : SUCCESS in 8m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1acd6655e2df4237b1af062c48e5a604 : SUCCESS in 1h 30m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0843f1272267475db9f4bf1f3466f52b : SUCCESS in 1h 47m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f02c7cf5d5f24a58812b2a8e88844858 : SUCCESS in 34m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/caee80bfcebc4e7688c7d40b4a24481c : SUCCESS in 31m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2834a171925348a99cc9a02252cdc1fc : SUCCESS in 32m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9ae1085bdba540c6afc13093e02b2358 : SUCCESS in 47m 15s","accounts_in_message":[],"_revision_number":1},{"id":"6eb6b94d82219b7e21f3ae5c1dc65d86dcf148e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 02:02:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d9d7cc3bd18f600139e43abc11fa07c5d89a0e95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 02:50:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9be25b81cf94c5283d5bffe60651ef6 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf012c2cda775d52abbd829f50fcdad07b29be5a","revisions":{"bf012c2cda775d52abbd829f50fcdad07b29be5a":{"kind":"REWORK","_number":1,"created":"2021-09-09 12:53:06.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/60/808060/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/808060/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/808060/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/808060/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/808060/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/808060/1"}}},"commit":{"parents":[{"commit":"f3b3e35af34667e2a7d718cc203f2ee03105c29e","subject":"Merge \"Fix race conditions between API and worker DB calls\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f3b3e35af34667e2a7d718cc203f2ee03105c29e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-08-31 20:20:44.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-09 12:52:58.000000000","tz":120},"subject":"Fix PlugVIPAmphora revert function in amphorav2","message":"Fix PlugVIPAmphora revert function in amphorav2\n\nThis function should call the unplug_aap_port function that takes\nDB models as arguments, but the subnet is not converted and is passed as\na provider dict. It triggers a \"\u0027dict\u0027 object has no attribute\n\u0027network_id\u0027\" error when reverting a load balancer creation.\nEnsure that the subnet is converted from dict to DB model.\n\nStory 2009157\nTask 43132\n\nChange-Id: I4bec17a484d0d39c9cca389421c63d378c36353c\n(cherry picked from commit b88154f086e5261d4a888598cce34e72ec1545b9)\n(cherry picked from commit c524eaf13e39ba84e1b6cb57579326f5925193ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf012c2cda775d52abbd829f50fcdad07b29be5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf012c2cda775d52abbd829f50fcdad07b29be5a"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
