)]}'
{"id":"openstack%2Foctavia~808100","triplet_id":"openstack%2Foctavia~master~I990b6af1ff880b25e54f6c41ae0c966007f3f098","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I990b6af1ff880b25e54f6c41ae0c966007f3f098","subject":"Fix MAX_TIMEOUT value for listener","status":"MERGED","created":"2021-09-09 15:25:04.000000000","updated":"2021-09-14 17:26:33.000000000","submitted":"2021-09-14 17:24:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808100","meta_rev_id":"111e4e6e334d18a04b9bdeee957b52afa355fd34","_number":808100,"virtual_id_number":808100,"owner":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-14 17:24:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-14 07:52:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2021-09-14 15:11:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2021-09-14 15:11:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-14 07:52:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-09 18:28:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-10 12:35:55.000000000","updated_by":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2021-09-10 12:35:55.000000000","updated_by":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"reviewer":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"state":"REVIEWER"},{"updated":"2021-09-10 12:35:55.000000000","updated_by":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-09-14 09:43:12.000000000","updated_by":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-09-14 09:49:42.000000000","updated_by":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2021-09-14 09:49:42.000000000","updated_by":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"0113cfb755e709c02b9de91b817d5c79de4ed292","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"date":"2021-09-09 15:25:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a6409bd3435b06c53a9d71bc530d5b917066f6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 18:28:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ba2b459513f4ffbb8ff58056e6b65e4 : SUCCESS in 13m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f832383e9c07472eb6ea98e0acaecb83 : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eb5a208f3874b8fa3083d344f204ded : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e551682fd2494300b5061b86dd1e1707 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4cd0ce3365b45fe8702354dea352f27 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53f5eb100ae44455b7fc87447fed66b4 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99e22d86b6f848b29386b280b307dafa : SUCCESS in 13m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/930b92bce9ea432c82992697fb04cefd : SUCCESS in 8m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/aa0ac3481ac74d6daaa06d1d286e78cf : SUCCESS in 10m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7da993fec119407d829df46a6499d99b : SUCCESS in 8m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/76ebf485f5644e8d965be064686c00c5 : SUCCESS in 10m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/69ba14da9b8141df9bc46d150ada10e1 : SUCCESS in 2h 47m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7042fc088664f108ab24a6bc8c3b978 : SUCCESS in 1h 26m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0d7445aaa7304547baf7f76bb73c59fd : SUCCESS in 42m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2e6b2c0ce2024647be4a94e912cb0094 : FAILURE in 1h 07m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b4ab1fb3c8e94cedbfed8a48696a41d1 : SUCCESS in 36m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/15bf696cff6d4699a4beb32beb306e58 : SUCCESS in 1h 22m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9c047ab465654ed199bdedef51ab21e4 : SUCCESS in 1h 47m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5900e7c6fedf45d7984cf3d7b6432ba1 : SUCCESS in 1h 31m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9f5f3d59d232421990dc248e5f792653 : SUCCESS in 1h 29m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ec24507978fc406a92583a2152ff1beb : SUCCESS in 1h 39m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ecb8f47dde1c45069ed0311d0c02a70f : FAILURE in 9m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e08324f81d0693acad71e913d0b896ce1181e639","author":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"date":"2021-09-10 09:20:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"29d05f9a257e5fb4b00ae75d5eb91f529d140c39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 11:57:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68835c7053e144de82d289d9092b99b7 : SUCCESS in 13m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3d91a45605146aca16baae9f2133d9f : SUCCESS in 12m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f53ff2095824b0b88a977e17f26570a : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d89d0b380524b8aa9b3735d72d0dec4 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8173849be9e472ba666e3a9237b749f : SUCCESS in 7m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b68e77d4c124e4fbfd6784ac1154241 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9400577df954090a3efc27a50c1d440 : SUCCESS in 15m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7ff9e1cbe31044e192ab168b2ad888f7 : SUCCESS in 6m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/91494fe560984b658b45e543a2576fdb : SUCCESS in 9m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c6a29b8f7f8c497e814c1794d67e769d : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef7f041b802a46d085e461988562b1d7 : SUCCESS in 9m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e58e730905a047bfab9817b4406e204c : SUCCESS in 2h 12m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f81101eafe54b5a9156d3a6d704d22c : SUCCESS in 2h 14m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2411585d9b1846f7900ebbcb1970595c : SUCCESS in 46m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/206699dc142843b795b70daaadfb0db0 : SUCCESS in 1h 11m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b3a7a1b889a4e588d92b7f4838b9d98 : SUCCESS in 46m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3f6dded20bd141b4b05ed1cc13b2b29e : SUCCESS in 1h 54m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d72ff666f9d54a29ac8d6d61e048c297 : SUCCESS in 2h 19m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/43aa74f4b37f44ed99c36f6393896342 : SUCCESS in 2h 33m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c8e1b2d26a61489ea91794717722c8eb : SUCCESS in 1h 41m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bc6e60e2072d43e29a541fd518363fa1 : FAILURE in 2h 22m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c0b86c89052e4259a4c962065c162b8f : FAILURE in 11m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ccdcc13cd48f7fad5baa40ef76b853a6634489ff","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-10 12:52:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)\n\nThanks for the contribution\n\nSome nit things in the test and the commit message\nCR-1 because we also need a release note for this fix.","accounts_in_message":[],"_revision_number":1},{"id":"8b93cd2c450a900eadc756f5942da3c63961813c","author":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"date":"2021-09-13 15:37:28.000000000","message":"Patch Set 1: Workflow-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"20fa1c9ce9b599863f480affc1a79dc4963986d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"date":"2021-09-13 15:37:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"afae9c6bbfc3594d9a276f9a17d08f57aad29e16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 18:20:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/159bf596af6a4baeb430a1a92ff5fdd8 : SUCCESS in 10m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/991ea45ba6ae48fa9a9e01aca59886f9 : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ee3fd25940b4c3797f11a87c2bcc8c0 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f5f87848b994e5288b77343fe30fdec : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02457a8be11e462da537d56fda939144 : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f4d1836bc834464bd7614d3007b1432 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feb033a97ebc413e990a3e2bd2db74ef : SUCCESS in 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e228e9557611432f84aa0fc71d54ff25 : SUCCESS in 5m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7ca010d6f1024fd78ac50ea8bd8af82a : SUCCESS in 6m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e044755031854204b10898e4cfa9d627 : SUCCESS in 7m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b813ef9429a4403db8d95f0d371cc6ba : SUCCESS in 5m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e84fc01840ac407f991533589ff84c38 : SUCCESS in 8m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/11d381f085b14a8f96a923e4ec2574a5 : SUCCESS in 2h 39m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5e9811e945e24eb98698d11ac527a94a : SUCCESS in 1h 46m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f72b706ed9b14297acab57563aaa5449 : SUCCESS in 44m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1b47e21032824b2197cb04746bd1efa3 : SUCCESS in 1h 04m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/568102ced9e848f282821929eddf4946 : SUCCESS in 42m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ccfcfa03a73944458d1405da37fba1dd : FAILURE in 1h 38m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7d82da92a83343999055c4197a0120d8 : SUCCESS in 2h 14m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/08e116f34d49454f99bdbddf85dc0c16 : FAILURE in 1h 52m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fe3cecbfc0494d9892452da59a662acb : SUCCESS in 1h 53m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/627162778b7647c0a5aac3248e10211b : SUCCESS in 1h 39m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5fb9b48c42cf4d44a5a37fe5033ea4dd : FAILURE in 11m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4e3f6b16d9ddb9f4c9ca389fce84bf25c3ff7ff7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-14 07:52:33.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2\n\nLGTM Thanks","accounts_in_message":[],"_revision_number":2},{"id":"16fad964bf1ad92697bd4b3b041eae033fbfcef8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-09-14 15:11:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nI would actually prefer to expand the database column size, but that would not be able to be backported. I think this is ok for now, and if someone requires a longer maximum timeout we can do the constants chance and database migration at that time.","accounts_in_message":[],"_revision_number":2},{"id":"192ee94379e824c8582472d9f3fa3f1a97a5155b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 15:12:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"34b5ef68cb6d851e438203657bd462010b9578e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 17:24:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66478b4cb7f440a79bf5fb6e4e99c143 : SUCCESS in 12m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/154835de5e41412d8db174fbc5746670 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efdb885a84fe41a7bf2d75461200bd03 : SUCCESS in 8m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6626d08ecdb24515aa97c781bb471f20 : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43a0658313fb4d078969bad8ed39e3e7 : SUCCESS in 13m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35a3dbc06d304fadb437f298ced80020 : SUCCESS in 6m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c312a2ae48894542965049e0c6e82621 : SUCCESS in 6m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/80b5e7b5803940d792a6cba14ce6969c : SUCCESS in 12m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f8df55e200c2460887c3fb91c502d0e3 : SUCCESS in 2h 08m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a1ce67d96065436fbdd7e7056240ef16 : SUCCESS in 1h 28m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2597d512a9774155a7e4843ccb433da3 : SUCCESS in 52m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f8a5d9f0162f4908af381208e8918c4f : SUCCESS in 1h 08m 38s","accounts_in_message":[],"_revision_number":2},{"id":"e312f8831141c464cc2fc412cdc6e75b59277c65","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 17:24:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"111e4e6e334d18a04b9bdeee957b52afa355fd34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 17:26:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b88a012410a48f393d03c742197736a : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5ad6aa8f8644648bf6716f6e5e4a542 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3a8f056306bba4502639bfc89752d8a48d694e7f","revisions":{"be6e30bd8c20de10b80869b371abb84be460c86d":{"kind":"REWORK","_number":1,"created":"2021-09-09 15:25:04.000000000","uploader":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"ref":"refs/changes/00/808100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/808100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/808100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/808100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/808100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/808100/1"}}},"commit":{"parents":[{"commit":"69d975d62315624bf15c02f5dfefe30c013e63cf","subject":"Merge \"Fix PlugVIPAmphora revert function in amphorav2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69d975d62315624bf15c02f5dfefe30c013e63cf"}]}],"author":{"name":"Roman Goncharov","email":"gadzhet007@gmail.com","date":"2021-09-09 15:04:12.000000000","tz":180},"committer":{"name":"Roman Goncharov","email":"gadzhet007@gmail.com","date":"2021-09-09 15:25:01.000000000","tz":180},"subject":"Fix MAX_TIMEOUT value for listener","message":"Fix MAX_TIMEOUT value for listener\n\nIn the database, timeout_client_data, timeout_member_connect,\ntimeout_member_data, timeout_tcp_inspect have an integer type with\na maximum size of 2147483647. Now MAX_TIMEOUT in API does not exceed\nthis value.\n\nStory: 2009193\nTask: 43248\nChange-Id: I990b6af1ff880b25e54f6c41ae0c966007f3f098\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be6e30bd8c20de10b80869b371abb84be460c86d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be6e30bd8c20de10b80869b371abb84be460c86d"}]},"branch":"refs/heads/master"},"3a8f056306bba4502639bfc89752d8a48d694e7f":{"kind":"REWORK","_number":2,"created":"2021-09-13 15:37:46.000000000","uploader":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"ref":"refs/changes/00/808100/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/808100/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/808100/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/808100/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/808100/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/808100/2"}}},"commit":{"parents":[{"commit":"69d975d62315624bf15c02f5dfefe30c013e63cf","subject":"Merge \"Fix PlugVIPAmphora revert function in amphorav2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69d975d62315624bf15c02f5dfefe30c013e63cf"}]}],"author":{"name":"Roman Goncharov","email":"gadzhet007@gmail.com","date":"2021-09-09 15:04:12.000000000","tz":180},"committer":{"name":"Roman Goncharov","email":"gadzhet007@gmail.com","date":"2021-09-13 15:36:05.000000000","tz":180},"subject":"Fix MAX_TIMEOUT value for listener","message":"Fix MAX_TIMEOUT value for listener\n\nIn the database, timeout_client_data, timeout_member_connect,\ntimeout_member_data, timeout_tcp_inspect have an integer type with\na maximum size of 2147483647. Now MAX_TIMEOUT in API does not exceed\nthis value and is equal to 24 days (2073600000 seconds).\n\nStory: 2009193\nTask: 43248\nChange-Id: I990b6af1ff880b25e54f6c41ae0c966007f3f098\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3a8f056306bba4502639bfc89752d8a48d694e7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3a8f056306bba4502639bfc89752d8a48d694e7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
