)]}'
{"id":"openstack%2Foctavia~808185","triplet_id":"openstack%2Foctavia~master~I387d9a3ad2197f6b59cd12db5f7643484c98fcd2","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-07-08 13:22:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I387d9a3ad2197f6b59cd12db5f7643484c98fcd2","subject":"Ignore status update on deleted objects in driver-agent","status":"MERGED","created":"2021-09-10 08:56:15.000000000","updated":"2022-07-08 13:26:02.000000000","submitted":"2022-07-08 13:22:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808185","meta_rev_id":"9c0a1442882b9cc607312b8c0748970609f0bece","_number":808185,"virtual_id_number":808185,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-08 13:22:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-09-13 07:41:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-10-26 07:27:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-08 10:58:26.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":1,"date":"2021-10-07 07:13:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-08 10:58:26.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-08 10:58:26.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-10 11:51:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-13 07:41:12.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-10-07 07:13:04.000000000","updated_by":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"reviewer":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"state":"REVIEWER"},{"updated":"2021-10-26 07:27:44.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2022-07-08 10:58:26.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"596ed53934864294132551ccc6e1af0851c2351c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-10 08:56:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b83d68be7d36dfcfcead744de00c1b8a7145df3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 11:51:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5622b6793f74062ac02137fc4a4b222 : SUCCESS in 11m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df70b6f9347e495bbb1e9da95fe92380 : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2e5503de0174c88b87ded452e66b33a : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4558783e7db4ba0944a64685d8cf83f : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff654173a1004e1d9d4bb2517f16387b : SUCCESS in 7m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41aa57f966144759a76a6aa1a6956140 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e708c40fb3a34f9f834c1bbff53c2a3c : SUCCESS in 17m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6b185e24918b4ec9bc1c928028e9a12e : SUCCESS in 6m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/857557c2adc043558f9e90e4fe2ea92f : SUCCESS in 9m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e13f18828038484aa5b34fb2c2a80b1f : SUCCESS in 7m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8819b2fe37504126a64e2289bf3485cb : SUCCESS in 8m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/84ce526028f942a28825ebfa0737607a : SUCCESS in 2h 20m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1301bccad40a43388992c01d00055f94 : SUCCESS in 1h 36m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7acd7a8ae7f14c938381962231abeb30 : SUCCESS in 37m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/12331b3ef7fa466f9d773c70a3f46c2a : SUCCESS in 48m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/535ec0435d6f47e08ea4953affdc58f5 : SUCCESS in 35m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/465b74a954294b2fbe65d7c61ae82cdf : SUCCESS in 1h 28m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/905ef70971f04baa9eb374bded437b3b : SUCCESS in 2h 04m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a6657222cf7546c083d162aa2bf957eb : SUCCESS in 1h 44m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1e2841d1de9b49299a4df6f417d56a25 : SUCCESS in 1h 24m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f55632bc4ae24c848e81ae9a70bb2376 : SUCCESS in 1h 16m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4d4ebc785c6c4cdbb2436c7ebbdf2a1d : FAILURE in 13m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3e7647214319800c19fcd7f843fb432df67d348","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-09-13 07:41:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7cbd017ad53977b383b317ca50b52f9b3e9728c1","author":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"date":"2021-10-07 07:13:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4365f2d37c6a4e12b8e231cbf9d6b5e6006ae0ee","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2021-10-26 07:27:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f6bb654943e086b22601f0205071e1a0c7f6b4f3","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-07-08 10:58:26.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a7450a3af2e3a366b7e23af98fe202e80e065010","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 10:59:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2dc31d06304781353aa294a9e43a6a1f06bac9c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 13:22:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6ee691901d0404fba5104d2ae1ba829\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/900c4e2b0fec4c189cddf71fa7e91e52 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58d1b23a781f4d9987af0bf0f09c3176 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0d978f1149b48489f767b1992ccbe61 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0cf6b1dd6464fb4a2cab758eb0dde99 : SUCCESS in 13m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/45583d05288d4497800bc215dc197991 : SUCCESS in 5m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/797fb1e90719482680b6dae83c5ae475 : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2f8da24004c24a4ba5f15d844a2c4fc6 : SUCCESS in 2h 08m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5bfc598f4616402da56dd889e00c991f : SUCCESS in 2h 08m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6b5d1e5372f043d7851eed5d77e6feac : SUCCESS in 43m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8161175c3b1c4692b92c3f615c03365c : SUCCESS in 1h 14m 13s","accounts_in_message":[],"_revision_number":1},{"id":"eb05b7ed48e3f6df14621535e104492cb5f464ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 13:22:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9c0a1442882b9cc607312b8c0748970609f0bece","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 13:26:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c7873a14d544063aa75f90def0fe30d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeb1870c2efc435ca8a60a449f40c6d7 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f5655b5eeec9bc1e59407b307233e9e6cdfc3896","revisions":{"f5655b5eeec9bc1e59407b307233e9e6cdfc3896":{"kind":"REWORK","_number":1,"created":"2021-09-10 08:56:15.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/85/808185/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/808185/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/808185/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/808185/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/808185/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/808185/1"}}},"commit":{"parents":[{"commit":"5dd7ad9ad8998c5285fe60d9ee4210f87ed45ccd","subject":"Add generic network interface management in the amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5dd7ad9ad8998c5285fe60d9ee4210f87ed45ccd"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-10 08:55:24.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-10 08:55:28.000000000","tz":120},"subject":"Ignore status update on deleted objects in driver-agent","message":"Ignore status update on deleted objects in driver-agent\n\nWhen receiving 2 DELETE updates on the same object, ignore the 2nd\nupdate. It avoids throwing exceptions because the object no longer\nexists in the DB.\n\nStory 2009202\nTask 43259\n\nChange-Id: I387d9a3ad2197f6b59cd12db5f7643484c98fcd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5655b5eeec9bc1e59407b307233e9e6cdfc3896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5655b5eeec9bc1e59407b307233e9e6cdfc3896"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[]}
