)]}'
{"id":"openstack%2Foctavia~808405","triplet_id":"openstack%2Foctavia~master~I8b1e3edfdc69f531b00cdef49121c45e944f25ca","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-08-10 18:53:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8b1e3edfdc69f531b00cdef49121c45e944f25ca","subject":"Catch exceptions on I/O in driver-agent","status":"MERGED","created":"2021-09-10 14:43:33.000000000","updated":"2022-08-10 18:55:16.000000000","submitted":"2022-08-10 18:53:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808405","meta_rev_id":"1697766f217c59d2381d8feef40b956c6403e9f9","_number":808405,"virtual_id_number":808405,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-10 18:53:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-09-14 07:29:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-08-10 16:45:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-08-10 16:45:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-10 18:52:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-13 07:55:40.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2022-08-10 16:45:22.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"67fd71b822d07184d226ab88465f2f9de834af8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-10 14:43:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"767a5bc33d4acf4d599e67730312527c3713c52b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 18:52:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3f35175441a48a9966a0a5444038563 : SUCCESS in 14m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83a8d201d16142c8aac307002398277a : SUCCESS in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ce246b1ed574dc19ac1f5d4c1e0f77f : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7091410cc4144d718a2f563f8f229795 : SUCCESS in 8m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43fc8aa11f144bb187f832bd225b4389 : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/523dcad55f1d4f7cb8d9a0644bc6e1c6 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbbed979e41d49759660749aac4d3c90 : SUCCESS in 21m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d59ef69c9ec74319ae0fb84af07c3856 : SUCCESS in 7m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4831febc3c254b77a5b2d1ef3591797a : SUCCESS in 12m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/877eb29d68bb42e288d5b5687e9f4b11 : SUCCESS in 8m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/42c0a6bcfe2847038c1c32b9c3fa335d : SUCCESS in 9m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0a5c155e76494149b542ca09f0fd4b5d : SUCCESS in 2h 44m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/91e1c3c33eab46f78eadd401995cd61c : SUCCESS in 1h 34m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/086e3849e4414208a829eccd7538085d : SUCCESS in 39m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d295eb8a6ac242629972036a8676d3eb : SUCCESS in 1h 03m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bdcea9c53cf64dcdbdc545a7c19d6379 : FAILURE in 41m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7400ae569f1241f4952a5caa5664d6b1 : FAILURE in 1h 35m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2219d0bf614140e79f34220b9a2041d0 : SUCCESS in 2h 19m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f56b3752dcb74b09a4eb1cf0339968c1 : SUCCESS in 1h 49m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5ccd66b1dd20426981c571a2875e6324 : SUCCESS in 1h 35m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9b95f260fff24d789ed1fd632d886fc7 : FAILURE in 1h 39m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f7ee0a44fc6b491191c02b1f4f3542c6 : FAILURE in 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f97d4e1ec9810f012a4bfad83bf036e0e36f4e2c","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-09-13 07:55:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d6923306d16de997431e3bcbf710d5763182f85","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-13 08:15:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6db78d97e13965ba0169ced8decd4e328ef60a6a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-09-14 07:29:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a0ab1f102d37e2e76c7a5d391dbbb0b6e6b9b496","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-09-14 07:29:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"674133dafb50d148cd9e44441c48f084a58edcef","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-08-10 16:45:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0520dfe65ec22a25af01752a8c946139b5cfa706","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 16:46:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3870729ab879951b9c9dd2e99ada51bab619d6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 18:53:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0946d9a3db2c4117b26afc403ba87b7e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/931a8c4d9f1b40eb872ddcf7f3317e07 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f395d179b24423b86c36753e6462afa : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a5db71125df411086dce55d2f5d5027 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e9b2541eaeb4ef6a7d61b8f1146b157 : SUCCESS in 12m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/64a4e1d634df463d81ae80ec9a7cda41 : SUCCESS in 6m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e01950b738a248a39e0416d565270726 : SUCCESS in 8m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aae01211136f457680d119bee79a45ba : SUCCESS in 2h 06m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d104e1dc49443d38a113d2c0100cf45 : SUCCESS in 1h 34m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/24572d1553894ff8ae627e54f983ec50 : SUCCESS in 45m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c206d39d3f4c467cb3c98e26f64f9680 : SUCCESS in 57m 07s","accounts_in_message":[],"_revision_number":1},{"id":"9d25b96dbc66ed4fed110f6b93eb9e618177f41c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 18:53:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1697766f217c59d2381d8feef40b956c6403e9f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 18:55:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5e44528c61f41ffbc1bbf109fedf747\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d10b863e2064d5ca082dff3a12b1238 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2886fc7224773dafbe44f95927476746bd7767e8","revisions":{"2886fc7224773dafbe44f95927476746bd7767e8":{"kind":"REWORK","_number":1,"created":"2021-09-10 14:43:33.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/05/808405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/808405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/808405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/808405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/808405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/808405/1"}}},"commit":{"parents":[{"commit":"5dd7ad9ad8998c5285fe60d9ee4210f87ed45ccd","subject":"Add generic network interface management in the amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5dd7ad9ad8998c5285fe60d9ee4210f87ed45ccd"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-10 14:15:00.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-10 14:15:03.000000000","tz":120},"subject":"Catch exceptions on I/O in driver-agent","message":"Catch exceptions on I/O in driver-agent\n\nIn the driver-agent, send or recv functions may trigger timeout\nexceptions. Not catching those exception breaks the driver-agent main\nloop, forcing the admin to restart the service.\n\nStory 2009032\nTask 42772\n\nChange-Id: I8b1e3edfdc69f531b00cdef49121c45e944f25ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2886fc7224773dafbe44f95927476746bd7767e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2886fc7224773dafbe44f95927476746bd7767e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
