)]}'
{"id":"openstack%2Foctavia~808544","triplet_id":"openstack%2Foctavia~stable%2Fussuri~Ib90f87f3e4e50e3118ee116614cfe31eb47e7282","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"Ib90f87f3e4e50e3118ee116614cfe31eb47e7282","subject":"DNM Testing octavia-amphora-image-build","status":"ABANDONED","created":"2021-09-11 08:00:45.000000000","updated":"2021-09-11 15:12:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ff503fcac0173e4f218541c7e33e95b7940a2752","_number":808544,"virtual_id_number":808544,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-09-11 08:17:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-11 08:17:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"74f8d01eb3bf1b0e79a52c6650e7dc82ffd878e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-11 08:00:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2523b0e5b618fd7e2328b9525804a88c21b5b299","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-11 08:17:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/779eb91c9f9e4b9094c35aa3f0eca21c : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de78e6abfd07404ab0ff5fcfddc7b791 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/825c8541073c4ebc8206a89a4ec29ff3 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb0c42acf3b24fb0b5f4c7cc5d782b8d : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20844e8be42c4e5985cc3a2109ad161a : SUCCESS in 11m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5abfa0fe4d9c47d1b7a2fd3f3b39b812 : SUCCESS in 5m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c9f5b9913ac94ae9b9056fcacf40a64d : SUCCESS in 7m 22s\n- octavia-amphora-image-build https://zuul.opendev.org/t/openstack/build/c232a39739684a4bb92b8af7c31f2eef : SUCCESS in 11m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ff503fcac0173e4f218541c7e33e95b7940a2752","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-11 15:12:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"243b2162704780f49973b03614501f90f7ac42b8","revisions":{"243b2162704780f49973b03614501f90f7ac42b8":{"kind":"REWORK","_number":1,"created":"2021-09-11 08:00:45.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/44/808544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/808544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/808544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/808544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/808544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/808544/1"}}},"commit":{"parents":[{"commit":"41b2aad609bcd758f44ce46f666261af4d655612","subject":"Fix periodic image builder jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/41b2aad609bcd758f44ce46f666261af4d655612"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-11 08:00:26.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-11 08:00:26.000000000","tz":120},"subject":"DNM Testing octavia-amphora-image-build","message":"DNM Testing octavia-amphora-image-build\n\nChange-Id: Ib90f87f3e4e50e3118ee116614cfe31eb47e7282\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/243b2162704780f49973b03614501f90f7ac42b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/243b2162704780f49973b03614501f90f7ac42b8"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
