)]}'
{"id":"openstack%2Foctavia~808586","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I3efa89992cc4a8e2645803dd867d7d2f6e39b966","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I3efa89992cc4a8e2645803dd867d7d2f6e39b966","subject":"Fix amphora image build jobs","status":"MERGED","created":"2021-09-13 07:08:40.000000000","updated":"2021-10-08 18:06:17.000000000","submitted":"2021-10-08 18:04:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808586","meta_rev_id":"8f17af26295131ea39bb60da685f8470771a0dde","_number":808586,"virtual_id_number":808586,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-08 18:04:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2021-10-08 14:58:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2021-10-08 15:06:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-13 07:08:40.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-09-13 08:56:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-08 14:58:58.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"4f645abfae0f2f932921f2abcad655d2d23614aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-13 07:08:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f00d9220f2310f6492b23917702e5ee91e9d1a2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-13 07:08:59.000000000","message":"Patch Set 1: Workflow-1\n\nWaiting for CI results","accounts_in_message":[],"_revision_number":1},{"id":"17233f79c93471ba64041532c2bfd2a4251b3627","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-13 08:29:54.000000000","message":"Patch Set 1: -Workflow\n\nJob passed in https://review.opendev.org/c/openstack/octavia/+/808588","accounts_in_message":[],"_revision_number":1},{"id":"5cb0adae541497b28d22e6372be6e82130c824a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 08:56:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b56886600e86457c960ee1176ac55f63 : SUCCESS in 22m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a31ce93df20433bbdddcca5ff8ae933 : SUCCESS in 20m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d6c3a5b26d444529be7616bea4c7a35a : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1dfd11df9c14cb9a707d8a322f82bd1 : SUCCESS in 15m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/411ad8b8b2b94abea443986b480bb488 : SUCCESS in 15m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeeaebcf778e477d903b96c224cbb593 : SUCCESS in 12m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8149db4288494fc0b6d1ab84c27f17ae : SUCCESS in 18m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bb0dbe18edfc4d3fb4b255d9901411b5 : SUCCESS in 22m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/53f3ea6bcf0a458c99cd7436aea75a01 : SUCCESS in 8m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5be03f1c87024ac29a53fbe887d3a0e9 : SUCCESS in 20m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/07fd70335f1a4bb48780ec1ec51141ac : SUCCESS in 1h 03m 25s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/675f04559c824541820e74f30e4ec231 : SUCCESS in 1h 09m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec320b1ce8ed4e2b8ec53b3ee1d3e963 : SUCCESS in 1h 24m 54s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3d7cb00fa5064d94aae0395fe929f39d : FAILURE in 1h 25m 22s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/169af351e3034267b63f6b089e77482b : FAILURE in 14m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2824240a47ba43f1b52bb729fbe96472 : RETRY_LIMIT in 10m 02s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/8687d20f24cf42ac82886271cf8598d8 : RETRY_LIMIT in 8m 55s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aca9b101bf6244debf97ff7c89ceffb6 : SUCCESS in 38m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/45f47f480fb94552afa5bc91561d2c84 : SUCCESS in 1h 28m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/140cb4d6e09342ed9240b704c0162b30 : FAILURE in 31m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0b9058d02749452aafa026fbfc2cf8e3 : FAILURE in 31m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/09372bc2ed28484f8fdd785a5fedcca7 : FAILURE in 14m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"17d9169dd33181abc5e8933c8e0730171adef04c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-10-08 14:58:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"913cd903696a457b24b2dfcb11a141ae9f7a97e4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-08 15:06:33.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f6f353319eb2decf6dac9e9e9a315aa3b6c99660","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 15:07:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b47667c1f9117ad8cc14431204118252a38366b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 18:04:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/060997c95db4443fa5a84a4decc49afc : SUCCESS in 22m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9dc26ce7b30545d685f7395752793461 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddb8d4a4e08f4b2589ef0882f67bc2a6 : SUCCESS in 20m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/231eb9b519b7412e86ee4f3c73a26a55 : SUCCESS in 21m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13ed2252d94e49c1b713e9342e4291c4 : SUCCESS in 12m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/931ae51cf0f3441ba5faaaea2d5fc6f1 : SUCCESS in 8m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/de31ad349e3048fe85a6adff6d37916d : SUCCESS in 21m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d384829b68f74f4986a1eb5568362e7d : SUCCESS in 1h 39m 03s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/08081520020a49a69cdda188d36f4c80 : SUCCESS in 2h 12m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ccfd41d5da9a42a5ba2287c07cc878a0 : SUCCESS in 1h 59m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e93af505877c441b9b79ed0026b63df681313259","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 18:04:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8f17af26295131ea39bb60da685f8470771a0dde","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 18:06:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01f8099f493145dfbd266ff9575b7e2a : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"150eafccf434ccb6692388cf8ce1e69ebcf77e2c","revisions":{"150eafccf434ccb6692388cf8ce1e69ebcf77e2c":{"kind":"REWORK","_number":1,"created":"2021-09-13 07:08:40.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/86/808586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/808586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/808586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/808586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/808586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/808586/1"}}},"commit":{"parents":[{"commit":"b1cc4467a916fea026454a7c3635a3681ab34253","subject":"Enable lo interface in the amphora-haproxy netns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b1cc4467a916fea026454a7c3635a3681ab34253"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-07-28 12:18:26.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-13 07:06:38.000000000","tz":120},"subject":"Fix amphora image build jobs","message":"Fix amphora image build jobs\n\nOur periodic amphora image build jobs started failing around mid June\nwith:\n\n\"The conditional check \u0027install_packages|success\u0027 failed. The error was:\ntemplate error while templating string: no filter named \u0027success\u0027.\nString: {% if install_packages|success %} True {% else %} False {% endif\n%}\"\n\nFilters have changed in Ansible 2.9 after a deprecation period.\n\nAdditionally, install python3-venv and set virtualenv_command (defaults\nto Python 2 \"virtualenv\") as it seems to be required now too.\n\nChange-Id: I3efa89992cc4a8e2645803dd867d7d2f6e39b966\n(cherry picked from commit 9df5f75d4913f966582290791c92a64b5e25acd4)\n(cherry picked from commit 20c0a88bf423a20cda6e9794f1a9123d15e700fc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/150eafccf434ccb6692388cf8ce1e69ebcf77e2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/150eafccf434ccb6692388cf8ce1e69ebcf77e2c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
