)]}'
{"id":"openstack%2Foctavia~808587","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Ifca01d91d8eb92115d56744f4963e91ac537dd8e","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"Ifca01d91d8eb92115d56744f4963e91ac537dd8e","subject":"Fix periodic image builder jobs","status":"MERGED","created":"2021-09-13 07:08:40.000000000","updated":"2021-10-08 18:24:13.000000000","submitted":"2021-10-08 18:22:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808587","meta_rev_id":"0e9f71a736168a69b49114ea869b922d8b85aa08","_number":808587,"virtual_id_number":808587,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-08 18:22:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2021-10-08 14:59:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2021-10-08 15:06:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-13 09:37:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-08 14:59:13.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"6d132713144e71a4bd51733b8ca25d6f26689fca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-13 07:08:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cce65a304352b223829d732522bf5e1ff227589","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 09:37:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/281eadfda16f41e5a2f3579c4a22e46e : SUCCESS in 19m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e899691fd2a4c3d8f2bcc05828623f6 : SUCCESS in 24m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7d89d4de30145c1a8f8347d417bf89b : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3b7a4373a084fbfba8bbd6e06c566d9 : SUCCESS in 15m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80c4ac59714a4dd0af03f812af2fdcb5 : SUCCESS in 17m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e70a0a5c0a8434d941af5fe40b6ec65 : SUCCESS in 12m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/950f49a5d33347358ca89a9332913ef9 : SUCCESS in 15m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bd6ba53090fd448fa51dc76980e47ee9 : SUCCESS in 16m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1ac6460571fe439bbb8a7e5c904f283f : SUCCESS in 8m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3c385829b5b04f32a9f55963abf94bb7 : SUCCESS in 16m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/66560509574a4af28878dc6bd9951a30 : SUCCESS in 1h 03m 38s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/915f7231bd0441779cabd3a3ab93dcf5 : SUCCESS in 1h 09m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0fbaa63def9743da9005ac985ab9b41c : SUCCESS in 1h 54m 26s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5364b14e7f1f467db4bce376045c3e41 : SUCCESS in 1h 26m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3b4b011ec84242469c5f15603336e310 : FAILURE in 14m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/82622f0036464ea7a1019ba9adee6c55 : RETRY_LIMIT in 7m 17s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/58bd75ab3b4d4f0c8e14d8aa18b22911 : RETRY_LIMIT in 7m 02s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f9abbb8d95b34cf5967af98640299e9f : SUCCESS in 37m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4655683720de4cecbf1b0d690241c6ee : SUCCESS in 1h 31m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/16bffae8923f4969993b9beed0639cba : FAILURE in 33m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c4d73d0f1c74449d88e83bb309bf8860 : FAILURE in 32m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/64babc1f90464e23b09597e7171c3557 : FAILURE in 14m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad61e3b35fc38f7475f9799744705d2d3411d281","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-10-08 14:59:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f1925c655a5cdba48fba4399ee7acba76b7d8984","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-08 15:06:28.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cf5ec619c3b9899aa671061d16aa8d57212c9a11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 15:07:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3542055884b811046c9cff104cf88c7d6d0f3656","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 18:22:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb1ed544d94c40f8809be178685400ef : SUCCESS in 21m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/38818439e0714f8a9fe0c43afc047912 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d2821e92c3545b986d0eab2473b1f34 : SUCCESS in 20m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57e1598a97c34c1e91d9f0aa879e1307 : SUCCESS in 22m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e89a890097c4722872a63b2a6342590 : SUCCESS in 11m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/69cf134e7aa049eb8ee58f661df0b1cd : SUCCESS in 9m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bea25621c20c44fc8793b7767f00113f : SUCCESS in 23m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/671445f1988343248d0dd840d62debd2 : SUCCESS in 1h 04m 35s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5bbedac5d9f14cbebd6986082918ddea : SUCCESS in 1h 10m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac9632e86a7441139fdbbec83dbfb426 : SUCCESS in 2h 23m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a5124cc2cb1c452f4a406356842e897b61ca2699","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 18:22:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0e9f71a736168a69b49114ea869b922d8b85aa08","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 18:24:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/121a244b2274486f8c2dd8b2aeafc6f3 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c78d1f192b77cfbf97ff83b9c6db6376e82f5f97","revisions":{"c78d1f192b77cfbf97ff83b9c6db6376e82f5f97":{"kind":"REWORK","_number":1,"created":"2021-09-13 07:08:40.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/87/808587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/808587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/808587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/808587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/808587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/808587/1"}}},"commit":{"parents":[{"commit":"150eafccf434ccb6692388cf8ce1e69ebcf77e2c","subject":"Fix amphora image build jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/150eafccf434ccb6692388cf8ce1e69ebcf77e2c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-12-17 08:34:41.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-13 07:06:47.000000000","tz":120},"subject":"Fix periodic image builder jobs","message":"Fix periodic image builder jobs\n\npublish-openstack-octavia-amphora-image* jobs started failing because\nubuntu no longer provides yum-utils package.\nNow dependencies have been cleaned up for the ubuntu job, and the centos\njob uses a centos node. The zuul playbook now works on Ubuntu and\nRedHat/Centos nodes.\n\nChange-Id: Ifca01d91d8eb92115d56744f4963e91ac537dd8e\n(cherry picked from commit d81a0556f575abbaeffbd812b859e489c3f035e4)\n(cherry picked from commit 80ff90e34eb0fcdb8c71da0a9f070e705f60e619)\n(cherry picked from commit 41b2aad609bcd758f44ce46f666261af4d655612)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c78d1f192b77cfbf97ff83b9c6db6376e82f5f97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c78d1f192b77cfbf97ff83b9c6db6376e82f5f97"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
