)]}'
{"id":"openstack%2Foctavia~808588","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I001214a9da38a8066b211e09ff27b397ad936150","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I001214a9da38a8066b211e09ff27b397ad936150","subject":"DNM Testing octavia-amphora-image-build","status":"ABANDONED","created":"2021-09-13 07:08:40.000000000","updated":"2021-09-13 08:29:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5dd393ef445b010500a84e41c32278c7cf076371","_number":808588,"virtual_id_number":808588,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-09-13 07:41:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-13 07:41:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3ac822293c3297451d035afaadc10844d0340ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-13 07:08:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4dd8adcb816c367735b7619b9a8ed6f7690079b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 07:41:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ede5b7ffc7744afb5f81f9af715b04b : SUCCESS in 19m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d7e44f33e77432581677d7d79fe0be5 : SUCCESS in 17m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/47b5466b7ef54c8fbbc60aeb6a4fc984 : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20dd2942a97247d4b922e5a5b4c45fdc : SUCCESS in 22m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87bb36673db3431c8399842b1e83e666 : SUCCESS in 19m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eea16a9f3a094955a08dfbf3b2f99c42 : SUCCESS in 13m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8685902df1084348b5bd0a0615cf1155 : SUCCESS in 14m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/dcc71ffcb5b348788da94a52c4075eaf : SUCCESS in 17m 27s\n- octavia-amphora-image-build https://zuul.opendev.org/t/openstack/build/0b5a1d2d46254aa9a10edcd8dbfa8c0e : SUCCESS in 10m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5dd393ef445b010500a84e41c32278c7cf076371","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-13 08:29:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6da7ff72a1d0d0121b4750256c9c278818b76867","revisions":{"6da7ff72a1d0d0121b4750256c9c278818b76867":{"kind":"REWORK","_number":1,"created":"2021-09-13 07:08:40.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/808588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/808588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/808588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/808588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/808588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/808588/1"}}},"commit":{"parents":[{"commit":"c78d1f192b77cfbf97ff83b9c6db6376e82f5f97","subject":"Fix periodic image builder jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c78d1f192b77cfbf97ff83b9c6db6376e82f5f97"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-13 07:08:16.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-13 07:08:16.000000000","tz":120},"subject":"DNM Testing octavia-amphora-image-build","message":"DNM Testing octavia-amphora-image-build\n\nChange-Id: I001214a9da38a8066b211e09ff27b397ad936150\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6da7ff72a1d0d0121b4750256c9c278818b76867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6da7ff72a1d0d0121b4750256c9c278818b76867"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
