)]}'
{"id":"openstack%2Foctavia~808977","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I990b6af1ff880b25e54f6c41ae0c966007f3f098","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"I990b6af1ff880b25e54f6c41ae0c966007f3f098","subject":"Fix MAX_TIMEOUT value for listener","status":"MERGED","created":"2021-09-14 17:48:00.000000000","updated":"2021-09-20 17:20:11.000000000","submitted":"2021-09-20 17:17:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808977","cherry_pick_of_change":808100,"cherry_pick_of_patch_set":2,"meta_rev_id":"237e116a4378890454d6988e995399e4552421cd","_number":808977,"virtual_id_number":808977,"owner":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-20 17:17:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-09-20 14:12:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-09-20 11:21:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},{"value":2,"date":"2021-09-15 07:05:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-09-20 14:12:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-14 20:28:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-14 20:33:38.000000000","updated_by":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-09-14 20:33:38.000000000","updated_by":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-09-16 19:58:11.000000000","updated_by":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2021-09-16 19:58:11.000000000","updated_by":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-09-20 11:21:11.000000000","updated_by":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"reviewer":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"state":"REVIEWER"},{"updated":"2021-09-20 14:12:13.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"91fc1e6386da14e44f05f9b2be170fc4eb2ddcda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"date":"2021-09-14 17:48:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"93b23df1ebf8cd0db732a623fa7e585a0eaa68dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 20:28:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ce4a502e8294b5092e4822a7a670ea6 : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4592d40c54ae40439c30faa41eed07eb : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fa42df014b34d35add78970606314c7 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60bdf226867a4247810496c6ca353d0c : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27eff63ce8ab4b24b434b05bb2d96377 : SUCCESS in 12m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd0579253da24a838d18ad749cb716fd : SUCCESS in 5m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9cc0c2819a3e4e7f83fd61212aedaf50 : SUCCESS in 6m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6c31091e9f394e3489420d821eee3f92 : SUCCESS in 11m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4e0df5549e524896b3d5d619f9182876 : SUCCESS in 5m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fe1499caae9e4880955adea6150b115a : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9e8a267bc31c4286bdfdeca7a3364c2b : SUCCESS in 2h 22m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5651f1f994fb47ee95ae206e687bf9f4 : SUCCESS in 2h 27m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b6f45b3473704cecb4e42e92ed2d333e : SUCCESS in 48m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5491088b910d4115b878237707560b6a : SUCCESS in 1h 13m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/513ed3273e144d1ab629a6c11d9c24bf : SUCCESS in 47m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a15aa438adb041868275af980016e6f9 : SUCCESS in 1h 25m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2773500526ba47dab5c36c3200437adb : SUCCESS in 1h 49m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aee0f553f20845dabdce3f81e5522d10 : FAILURE in 1h 18m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0195d8cc16454bab8b5c978aed849702 : SUCCESS in 2h 20m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1dd19b8b43654d33acd809e4e85c3be9 : TIMED_OUT in 2h 35m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/115f0d41e0594921be6b0adb281f0806 : FAILURE in 1h 36m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6e9c02005e9446399811ffeec2a9f39e : FAILURE in 37m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e10d8dd3bd881ec258a2d385cb890829b6a48594","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-15 07:05:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"64f3c17bf335f0e80b946bbd543ba9bd4584f4f3","author":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"date":"2021-09-20 11:21:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b0936cb727e440919d4a475ac08c31c881cd2818","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-09-20 14:12:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db037cadc461868baabfd4d72174264d6c5e168c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 14:13:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58f28a631bc2a8165ebaee39a4d8225f67c91a28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 17:17:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce6cfe1d2acc4b7ea0566dfcdd813683 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa451093ce314e0a9041eb32272e14ba : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c34751bdb314fb4815e25f22d8a30c0 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86c90a12da7b471cbb72bafb431f2959 : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1837724408843d7946b97916f7043b8 : SUCCESS in 5m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f72e2d70cf3143aebfba464cfd936057 : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b890b17afa744cfaa679f9d01ea55958 : SUCCESS in 8m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5364929d8cff4b78a62f0694dc484e22 : SUCCESS in 1h 34m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f5edecee3f4a4e83a60263772b85572b : SUCCESS in 2h 02m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d91c4a2c08f6409cb84ecccda1ec4136 : SUCCESS in 44m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9bb24067610044b7acf1486e45fdcb4e : SUCCESS in 46m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd9c5ed403c44a9ab5faf47e24c5ce1e : SUCCESS in 42m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/623032603144460abaeb7e9740f0d6f2 : SUCCESS in 52m 30s","accounts_in_message":[],"_revision_number":1},{"id":"459f0e33a6342d838554be9afa9b45d301c4fb7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 17:17:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"237e116a4378890454d6988e995399e4552421cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 17:20:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18e08772e2604af884eead9b0d515e76 : SUCCESS in 1m 27s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/986a3b6c80a348b78597eeb7c06450d2 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b94493eb2abec3dceba67ed54237018c133172c8","revisions":{"b94493eb2abec3dceba67ed54237018c133172c8":{"kind":"REWORK","_number":1,"created":"2021-09-14 17:48:00.000000000","uploader":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"ref":"refs/changes/77/808977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/808977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/808977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/808977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/808977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/808977/1"}}},"commit":{"parents":[{"commit":"f3b3e35af34667e2a7d718cc203f2ee03105c29e","subject":"Merge \"Fix race conditions between API and worker DB calls\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f3b3e35af34667e2a7d718cc203f2ee03105c29e"}]}],"author":{"name":"Roman Goncharov","email":"gadzhet007@gmail.com","date":"2021-09-09 15:04:12.000000000","tz":180},"committer":{"name":"Roman Goncharov","email":"gadzhet007@gmail.com","date":"2021-09-14 17:48:00.000000000","tz":0},"subject":"Fix MAX_TIMEOUT value for listener","message":"Fix MAX_TIMEOUT value for listener\n\nIn the database, timeout_client_data, timeout_member_connect,\ntimeout_member_data, timeout_tcp_inspect have an integer type with\na maximum size of 2147483647. Now MAX_TIMEOUT in API does not exceed\nthis value and is equal to 24 days (2073600000 seconds).\n\nStory: 2009193\nTask: 43248\nChange-Id: I990b6af1ff880b25e54f6c41ae0c966007f3f098\n(cherry picked from commit 3a8f056306bba4502639bfc89752d8a48d694e7f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b94493eb2abec3dceba67ed54237018c133172c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b94493eb2abec3dceba67ed54237018c133172c8"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
