)]}'
{"id":"openstack%2Foctavia~808978","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I990b6af1ff880b25e54f6c41ae0c966007f3f098","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I990b6af1ff880b25e54f6c41ae0c966007f3f098","subject":"Fix MAX_TIMEOUT value for listener","status":"MERGED","created":"2021-09-14 17:48:38.000000000","updated":"2021-09-20 20:19:36.000000000","submitted":"2021-09-20 20:17:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808978","cherry_pick_of_change":808100,"cherry_pick_of_patch_set":2,"meta_rev_id":"fd150d10f3372be495f7f7a76fa1b4f277cb9130","_number":808978,"virtual_id_number":808978,"owner":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-20 20:17:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-09-20 14:12:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-09-20 11:21:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},{"value":2,"date":"2021-09-15 07:05:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-09-20 14:12:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-14 20:16:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-14 20:33:51.000000000","updated_by":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-09-14 20:33:51.000000000","updated_by":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-09-16 19:58:14.000000000","updated_by":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2021-09-16 19:58:14.000000000","updated_by":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-09-20 11:21:27.000000000","updated_by":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"reviewer":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"state":"REVIEWER"},{"updated":"2021-09-20 14:12:06.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"7cc1e1e5d2f22f180e6112e1f42416aa68e600da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"date":"2021-09-14 17:48:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e2baf8db34e9dd1ff401f0a15305ce53dddefea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 20:16:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/033d4f0bfdd54b3a9375bd7b2e0b583d : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c45f7f3730240a8bc85c214115dd6ad : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/774e092c45e14af09ffb486c86c6a923 : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc4c8dcf923145168c07e9537546b140 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c83d6088ddb14be5a97034ca7c137be2 : SUCCESS in 11m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/524c0c014c1f4668a58fa43e1076800b : SUCCESS in 6m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/99ea9574cace488e9c9582c3311c2d1b : SUCCESS in 6m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/909dab969a05448e98b72e634ceba9de : SUCCESS in 8m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8eccf63ae26e4d2898267add4407521a : SUCCESS in 7m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/49d2bf602ce14df58d6b574542b3c904 : SUCCESS in 1h 40m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/262668a461294549bb6852ad80cbd96b : SUCCESS in 2h 19m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/af0e5cce37a24ba2b7aa926c45cafc32 : SUCCESS in 51m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/12ac944f847b452dbadae9ee9d1541d7 : SUCCESS in 52m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/03c87e1612da4e92bf20200478692af2 : SUCCESS in 1h 10m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f48b238c7ee04ba1ab4c2ec6a2b0d689 : SUCCESS in 1h 45m 08s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6c0420b12fec422d994968da715dc457 : SUCCESS in 2h 08m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7657a9ae0029420f8ad4fe26f3a6ed89 : RETRY_LIMIT in 9m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f05c6fc192624f3e88a470a6f747e895 : SUCCESS in 2h 25m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"246af760411b0ce9f39fa72efad76601042ccca8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-15 07:05:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0f0f0f84170902de8bf8d4c497ba67430404b2a0","author":{"_account_id":27256,"name":"Vadim Ponomarev","email":"velizarx@gmail.com","username":"velizarx"},"date":"2021-09-20 11:21:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"215ded8cda03eaff66bded2e9f4cd0a5bc6e8cf6","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-09-20 14:12:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b88371e620fff0edc94ee07311577caa7a59ca26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 14:12:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cae8801441c34d6d1266033ca2b5c5b6e8cff7aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 15:11:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67faf1f1e41e4fc5a4558b054a393fb8 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af4a478b074443dead48f912c88e3514 : SUCCESS in 6m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ed8608040544d8a8bf14d3fb035337b : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4c1781a819c4901914d917b565e1180 : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af931c54118e411d918cf2d22f91b32e : SUCCESS in 5m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0f7612540ade44f3850c2a95824f9c54 : SUCCESS in 7m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4679796d65c24feba3dc7c80f13a1d86 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/98a917d45d994894a9941ff32cfacabf : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d12127e0e87648c6869a15522a5519c2 : FAILURE in 51m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f49bcfda609b42bc96aeb732e0008b15 : SUCCESS in 47m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6a4f3d94d8ab4713a3630eec8fd1205b : SUCCESS in 41m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e8369cabe9e5497aa5ba42cb4bfe1091 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"124b01c48c6427fba2ad4970392031aedd0c6882","author":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"date":"2021-09-20 15:34:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7c0384e79ecc01a900061d574230f63ba926b546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 18:00:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4461c84e11934d37b0239fa206fdb0d5 : SUCCESS in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a67a3e311fa34540aa39dadf34470a8e : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3d6234d13ec4e10ae09a5e4f5ec34c3 : SUCCESS in 10m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b28214a1d59484994bfcd1ce2bebba5 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b90bda25811f4c4bad559eaae8752c4a : SUCCESS in 15m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cd0b31d1717421a808ff6bbd063d3da : SUCCESS in 6m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/46c692d2957743409b2bd8153a6005e8 : SUCCESS in 9m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3daf6267912d46bc98dda27fe467b55c : SUCCESS in 8m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/93ea537ec6bd4ac09bbe6429d5778b1f : SUCCESS in 11m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1a9b8c0c6eed4998902e0bff9c90141c : SUCCESS in 1h 34m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/098bd546dbb44143b2e0d5236da832ea : SUCCESS in 2h 21m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/90fba0f5051447f6baf21a9057e06059 : SUCCESS in 44m 27s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2abcdff661624d3a8ee4c81b50c3ba11 : SUCCESS in 56m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fdaae8adaeaf4f0781cee870bfcd9ec8 : SUCCESS in 45m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7e8d22078b5b43de9bc59a84c4424ccf : SUCCESS in 1h 06m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b60a22a03e36461e91547f801bfc4eed : SUCCESS in 1h 45m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ce9a36dff46242df9ee8e86658de0f1c : RETRY_LIMIT in 9m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/256c79eaf95241c48b13a05522228c34 : SUCCESS in 1h 58m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a90d9f35bb1f8b614c74999a68990d8c18fff52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 18:02:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7edcc393a64211e549c3025f162e9eaccc423167","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 20:17:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc513b5355c84085a76b323b795f0e51 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a1da1604b7949878942b6d2f41fd343 : SUCCESS in 6m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a3aaf2c18ce448698fda3d7205ae1ec6 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19d3081648d746439530b26914cd0a28 : SUCCESS in 12m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a32fa259f3f1446da1d78af560bfb265 : SUCCESS in 5m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c2221add3d1140388b7327bf615133f7 : SUCCESS in 8m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/56c2b30543c54dd3a0564e22fd66d4e5 : SUCCESS in 1h 11m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c5f80874e91477bb4dcec79c85db218 : SUCCESS in 2h 08m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/18081023dfe1453ca049ec5d8a838f13 : SUCCESS in 43m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6cf1dc33f4be44abb2b138b31879c98f : SUCCESS in 44m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8790b83b109f4a9a8c6500587a309fa0 : SUCCESS in 46m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f5e2f721a863406ab8571a95b83310c2 : SUCCESS in 1h 12m 42s","accounts_in_message":[],"_revision_number":1},{"id":"0b706eba8274c1797e9493786c1314d2e20672f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 20:17:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fd150d10f3372be495f7f7a76fa1b4f277cb9130","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 20:19:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29dd153bdd95405b8c71e11ea2433e8d : SUCCESS in 1m 34s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb01e0b5c67e4ffd8f045afdee9c6e31 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a25a643e2cb544bd9b4993a126a74ecc482ac2d","revisions":{"9a25a643e2cb544bd9b4993a126a74ecc482ac2d":{"kind":"REWORK","_number":1,"created":"2021-09-14 17:48:38.000000000","uploader":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"ref":"refs/changes/78/808978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/808978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/808978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/808978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/808978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/808978/1"}}},"commit":{"parents":[{"commit":"d79c80c39c6953768c267abdba0c38a7374d6e2b","subject":"Fix race conditions between API and worker DB calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d79c80c39c6953768c267abdba0c38a7374d6e2b"}]}],"author":{"name":"Roman Goncharov","email":"gadzhet007@gmail.com","date":"2021-09-09 15:04:12.000000000","tz":180},"committer":{"name":"Roman Goncharov","email":"gadzhet007@gmail.com","date":"2021-09-14 17:48:38.000000000","tz":0},"subject":"Fix MAX_TIMEOUT value for listener","message":"Fix MAX_TIMEOUT value for listener\n\nIn the database, timeout_client_data, timeout_member_connect,\ntimeout_member_data, timeout_tcp_inspect have an integer type with\na maximum size of 2147483647. Now MAX_TIMEOUT in API does not exceed\nthis value and is equal to 24 days (2073600000 seconds).\n\nStory: 2009193\nTask: 43248\nChange-Id: I990b6af1ff880b25e54f6c41ae0c966007f3f098\n(cherry picked from commit 3a8f056306bba4502639bfc89752d8a48d694e7f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a25a643e2cb544bd9b4993a126a74ecc482ac2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a25a643e2cb544bd9b4993a126a74ecc482ac2d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
