)]}'
{"id":"openstack%2Foctavia~808986","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I9dbdb71e9b94bbcc75766a8687a996d5358f3381","project":"openstack/octavia","branch":"stable/wallaby","topic":"story/2009117-stable/wallaby","hashtags":[],"change_id":"I9dbdb71e9b94bbcc75766a8687a996d5358f3381","subject":"Fix duplicate SG creation for listener peer port","status":"MERGED","created":"2021-09-15 03:06:36.000000000","updated":"2021-09-21 11:04:31.000000000","submitted":"2021-09-21 11:02:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808986-story/2009117-stable/wallaby","cherry_pick_of_change":804485,"cherry_pick_of_patch_set":7,"meta_rev_id":"504b6331777e71f45bc2ff9f2f4aac10e405c73f","_number":808986,"virtual_id_number":808986,"owner":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-21 11:02:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-09-20 14:14:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-09-15 07:03:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-09-20 14:14:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-15 05:12:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-15 07:03:02.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-09-20 14:14:45.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"344ce952f0571544c3b00e18c93ac7f650d82e0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-15 03:06:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9efbb1bb233c29c037e9d1e821c547b3df361e7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 05:12:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1892c66e7fd642dbaf1cbc30bcf57422 : SUCCESS in 10m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e69bea026941435380695d6599adcc17 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6cbcddbbc8f4fc9b49497ee2e4cf624 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c88cc80767894e69b0695f5870826b81 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/160945d7d060411da8f8073c168ebce3 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b7132bf42254100a0b12762a9af614a : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9ab3a4f4e734240a6eba4e1e90289fe : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfc9e9c156694e8886078351ed0bc044 : SUCCESS in 5m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b4193e2141f34925aa9539380a45b4e3 : SUCCESS in 6m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3da3d051452b444fa15591fcc04e3a66 : SUCCESS in 8m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d92703ac10324ddca013e6099634a8f1 : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7e2ecf1f502c4b0baf29063677a27896 : SUCCESS in 7m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eb8c74454d994cd1921f6a977695ad9f : SUCCESS in 1h 30m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bddad48a2684460b94893323a991fa15 : SUCCESS in 1h 25m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/36d81f4ce3a84c3aaea5e463684b37bc : SUCCESS in 44m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2b520e5348864d2a83d39a817629dc6f : SUCCESS in 33m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4c7967e57c749e383c171a4bdf9a2b4 : SUCCESS in 35m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a3190bc6e1e84d4fb1f6e3ad7635b415 : SUCCESS in 1h 17m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a6d6d0cae18a42139b02964002b9283e : SUCCESS in 1h 56m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8ff6c2130f7642c9afe7e8c9d0d000de : FAILURE in 52m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/aeea3e563ff64a6fa8e76ca49414a0a6 : SUCCESS in 2h 03m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d4bce1f6cd7648148c0f91aa4a3a7016 : SUCCESS in 1h 56m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c5b17f39c6724ef6a264a9cc653252c9 : SUCCESS in 1h 52m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fcfcb7446a2c4bdcb3ea37326c138944 : FAILURE in 9m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"469bdee440e41a38baa3d776fa63c70b3db8a5b0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-15 07:03:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"550da8387a90cc924adc30e400789502a34e4972","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-09-20 14:14:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"15a49c3fc942f200f9193620bfcee1737b0e9e2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 14:16:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e41d2fab648f5ee20d769eb44a2fe41f188f3938","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 17:38:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be8c6d0cfe674ba28b20d7349197e1a9 : RETRY_LIMIT in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/635bb5f203d64a19b747f6987cccf159 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76da6cb1e46341cd833f62c697f6b939 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/010a44f631cd4062a8354b2b8e2ad59a : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e94c96cf85747f2bb00028ce4274e70 : CANCELED in 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ad1decabfaa40809bf7f93de815b996 : SUCCESS in 5m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/377af10d217e4588bf5d340bf8ca5463 : CANCELED in 8m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a6803699a0874393b47a650a7cf8d472 : SUCCESS in 7m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/410409040eb442998f67771d52a87b4e : CANCELED in 12m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/76f48a7ee1164c7ba9d8c5a3a9b4a408 : CANCELED in 12m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/118a7a6fc3be4ff78a461ec294e8b08e : CANCELED in 11m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/da5011acc9514ee5bade47410bc8b613 : CANCELED in 11m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/215a2e6974474f0b8cacee103068e6d5 : CANCELED in 11m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b7e53158687449e5861855d081e6a25e : CANCELED in 14m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5a6d7e24c1059b3ac392dfeb60d2c1b71c887b65","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-21 02:54:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"317a233ca54ad9c67e3e269b75dab35462001412","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 05:09:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab6b46467bd6475db4e5ec7c93a7d85b : SUCCESS in 10m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af7da7ec7b5b45959d6eeb48ac3dc061 : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2f29effa12d47ed8d203816e95ec9b6 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e86e9be6e1a94ffe98c54a88b8a70591 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce987d1990a442daa2a6a4919ffec97a : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d882ab2c909e49ecb36fac55d716a7a8 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/670c24f982fd49608a1ef25685738c3a : SUCCESS in 12m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b15e7fb7dc774f36a22e7bd9ad262b58 : SUCCESS in 5m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5a9ec40055d544ae87022390bd10558b : SUCCESS in 5m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9942b943e4344994b32917e24c35dcae : SUCCESS in 8m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0ca320a0558049cfa9b6c8f3141f1de7 : SUCCESS in 5m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bf5d9b62b8534c7abeaef812fec29167 : SUCCESS in 7m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4737b35a9303477898c8b4032ef8d6eb : SUCCESS in 1h 17m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f0264634ed5d4e0a921e45bc362e81b5 : SUCCESS in 1h 43m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e656305ed7764c48996b73203b7ba6c3 : SUCCESS in 35m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/475eabe0a251401ca14311959498116d : SUCCESS in 39m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e2a93b82386c4fdda1da4cc117a867b3 : SUCCESS in 38m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9b8e7ef624484fcbbb2c3573d2d392f0 : SUCCESS in 1h 00m 44s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/552b11bcfa3b441d9541540266e42c94 : SUCCESS in 1h 49m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/256fe2c7d937430e8035768fc3a9ba97 : FAILURE in 35m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f83d80af8fd3459aabfa2032da9b913e : SUCCESS in 1h 53m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/235a9cafd59d4d10b7118eff83fb9636 : SUCCESS in 2h 11m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fcb63bb9bcb5467e8d7b789df9261b15 : SUCCESS in 1h 36m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f8d48f8b64fc49f39b0c862586204268 : FAILURE in 9m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73ba8837c7a6b6d835c6195dd10f8aab9be9c8e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 05:09:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6271650c7f4ab50c526e009da1011fc60939380e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 11:02:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c90215a501343938fd05b4965754b0d : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f76448a99fc4a3292e2cd32a363253c : SUCCESS in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67f9f53dc60e4a2baec76453ff20de50 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3997992b316045a8a5006d12688e7087 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f3e7f989be247778bcbd42f8dc95032 : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1369daa12f6647cd9f79976e42a9f8f1 : SUCCESS in 6m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f3bb3f919bbf4426ad9db3bbc10b6f17 : SUCCESS in 9m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d6968c36b9084bc59892f478da63877a : SUCCESS in 10m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8c0cfd45f5dd43dcbe31c09a34907957 : SUCCESS in 1h 32m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2dd3d8d493934a6ebc82693efc973fc0 : SUCCESS in 1h 28m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4825ea2153cf40bca0d0a7f830c166c7 : SUCCESS in 35m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/84fac86816c347c39fcf0c8abbaee287 : SUCCESS in 34m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe438c0151124204b7346a3ee64a9c20 : SUCCESS in 33m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/40556afe704d4ea995036823c0c9cac2 : SUCCESS in 47m 44s","accounts_in_message":[],"_revision_number":1},{"id":"9e2a296df969b8aeb7e7611d8899810589938436","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 11:02:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"504b6331777e71f45bc2ff9f2f4aac10e405c73f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 11:04:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/657b13edce3c4bb280aa30c12d9b4134 : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aab4755c569444c0bc7829f72a812305 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e692201fad32ba1e4afc33463df567bd70fcdab0","revisions":{"e692201fad32ba1e4afc33463df567bd70fcdab0":{"kind":"REWORK","_number":1,"created":"2021-09-15 03:06:36.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/86/808986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/808986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/808986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/808986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/808986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/808986/1"}}},"commit":{"parents":[{"commit":"f16f72cca2a1fec54896cad72100607f2acc9b7f","subject":"Add caps for pip-extra-reqs/pip-missing-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f16f72cca2a1fec54896cad72100607f2acc9b7f"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2021-08-13 06:24:59.000000000","tz":330},"committer":{"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","date":"2021-09-15 03:06:36.000000000","tz":0},"subject":"Fix duplicate SG creation for listener peer port","message":"Fix duplicate SG creation for listener peer port\n\nIn cases where the listener protcol port is same as the peer port\nand allowed_cidr set to 0.0.0.0/0 explicitly, the listener is not\nprovisioned due to duplicate security group creation for peer port\nwith None as remote_ip_prefix. Neutron SG defaults remote_ip_prefix\nto 0.0.0.0/0 if not specified or None and hence the error SG rule\nalready exists.\n\nRemove the duplicate entry from the updated_ports.\n\nStory: #2009117\nChange-Id: I9dbdb71e9b94bbcc75766a8687a996d5358f3381\n(cherry picked from commit 151a94321074b1eda53d1a582fdf26942db85afd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e692201fad32ba1e4afc33463df567bd70fcdab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e692201fad32ba1e4afc33463df567bd70fcdab0"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
