)]}'
{"id":"openstack%2Foctavia~808987","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I9dbdb71e9b94bbcc75766a8687a996d5358f3381","project":"openstack/octavia","branch":"stable/victoria","topic":"story/2009117-stable/victoria","hashtags":[],"change_id":"I9dbdb71e9b94bbcc75766a8687a996d5358f3381","subject":"Fix duplicate SG creation for listener peer port","status":"MERGED","created":"2021-09-15 03:06:52.000000000","updated":"2021-09-22 07:32:39.000000000","submitted":"2021-09-22 07:30:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808987-story/2009117-stable/victoria","cherry_pick_of_change":804485,"cherry_pick_of_patch_set":7,"meta_rev_id":"d6dc6ba870db8e26966c4550434ede572d4aaede","_number":808987,"virtual_id_number":808987,"owner":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-22 07:30:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-09-20 14:14:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-09-15 07:03:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-09-20 14:14:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-15 05:13:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-15 07:03:30.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-09-20 14:14:38.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"408cee3b0c8a1506f7481bff0f352e373045c8e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-15 03:06:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7fc394395e8f73f3372b202962217e00cc3f92a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 05:13:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6cda716d0ee54751abf78d720030b54c : SUCCESS in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/866eebba6eb74be4bf36f1fbcad3ceb4 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/066705a6091a4ac6994b70fbad3024cb : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24c9142c93ee4432953c0b9a1a867da0 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d0228bd94154dc1bd4f8d1107e4426c : SUCCESS in 12m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b0366e1695548aaafee0430f3e57e01 : SUCCESS in 5m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/858138dc4f8549ef894b7948c1a074be : SUCCESS in 6m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ae1bc4a4d8274228894ae364a7a4d2c6 : SUCCESS in 8m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d72d0f1ea38f449d90d41288a7d54ec0 : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2e60d04200b44016b57ff90ebeb2ddd1 : SUCCESS in 7m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8403471a343143b5848d5dcd87606b18 : SUCCESS in 1h 29m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28e0e2b02e4e48f18736dcf6b07842d2 : SUCCESS in 1h 26m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/84e45728509e4d209df58b3fc8e571d8 : SUCCESS in 38m 23s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/be49e32b93dd485e96694731b6af3abd : SUCCESS in 33m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6da674ca789b4eb9b2174da71cb03b36 : SUCCESS in 33m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a319f4bd228b4ac797acbab278108489 : SUCCESS in 54m 19s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d80f2d256b1c44298019cf6a6f2eb4d7 : SUCCESS in 1h 43m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bf373508e2714f62a60e79f5a4ea7ccd : FAILURE in 33m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a77aec3d4ac940358917324ff5a50e70 : SUCCESS in 1h 35m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5c130f6dc31b4734966bf6fcfdeece2e : FAILURE in 1h 51m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bdc48251299744f78d75dee13de76cc3 : FAILURE in 1h 45m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8a05c3e670fc4898897c1c6da4a8d66b : FAILURE in 29m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"037a4ddef4c4bd8ded534a8d691826c0f3430a13","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-15 07:03:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2535ea727b0eeed0679d79a94058d3172a28d4af","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-09-20 14:14:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8a22d7c74ef4253022932490cdbe206a45ce0c9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 14:16:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8b876e8d3b2ab37d142bc3f7c6ddd753e9f49ec7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 17:38:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08732c38da8648c2aef5692bcf724dfd : CANCELED in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a9464b8f592458ca105397931207009 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66b46b63ce3b417187b8d60affcb7526 : CANCELED in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6c4832679ea4acb87d52731c81ba102 : CANCELED in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/419556fc9b6b4551b80e58a26ae16789 : SUCCESS in 6m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e7fd4e34a6444559ade725adf69cf12f : SUCCESS in 6m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0e2b97e408e941299a0c2bfbecbb2b55 : CANCELED in 7m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e2c51a1947184152a5b6a88b674b3a5c : FAILURE in 9m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/63f2713e42114e3d83ef00b1515bdcad : CANCELED in 9m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d9416ed283074ddcbf3df2fbd1d536d5 : CANCELED in 6m 56s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/274d110ddfde452a98517abd81e5ec12 : CANCELED in 8m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3fa4201099034ab18232e8fca4b5e2f3 : CANCELED in 10m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/af03a0df430045fcad4b5b78e6735d58 : CANCELED in 8m 46s","accounts_in_message":[],"_revision_number":1},{"id":"811f0701a285b423cabe76fad73f67d50a108811","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-21 02:54:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"958444316f337e258103d479ae8ca8a19a28595c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 05:08:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd92e5b7a82e4e178a244b73a8f8a861 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10c09fbb0492401d9854bcd6cb8e7cd3 : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ae1169d69a04ca088ac71e4eb078eba : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d73bbc0eee44822b81cb18a91c19d05 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/507fd01ce6ce44b88ecac309d679b700 : SUCCESS in 13m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07f08b4f8ed7499289184d6430472f8b : SUCCESS in 6m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7967b98ade3d46f28a6a4b9f4558aa5d : SUCCESS in 6m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7a613055abf846c89448183f061dc4c4 : SUCCESS in 8m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/db169649471245b0be6008693c9aa0ad : SUCCESS in 6m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4ab07693ffa74adaa26a1564c00c6ef4 : SUCCESS in 10m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c1afcd0a20a34cc290a051f4e87010a7 : SUCCESS in 1h 46m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e59e510ade04ca08d11dec77b433392 : SUCCESS in 1h 49m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/24fe631d17cb498088363fdca45fb4ff : SUCCESS in 52m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a084a182ff9f4c1b911cc32980d0bde1 : SUCCESS in 37m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/23f427f3b2744c27a512af02ae1f7690 : SUCCESS in 37m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b3e3f7117c7c4f75864e5bb5e06037a5 : SUCCESS in 1h 15m 18s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c0c33af6ebd8402ba2bdc6e9f055a058 : FAILURE in 2h 10m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0d304f8682034d0f9e328c435a575eb6 : FAILURE in 46m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5687669ba95744488f408f5f2dacd11f : SUCCESS in 2h 00m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b398510a2c0e4eca9425b8fae1623cf8 : SUCCESS in 1h 56m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9bce3beddcce4e34a56fde8ff23cc617 : FAILURE in 1h 37m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/57e599ebfbf24529ad6967d50c70ce2e : FAILURE in 24m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92a2cbbbef7b2e2e86c6a97900411ce0996be781","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 05:08:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c9895ea07ff60c0db294b9e085a25f02b4963e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 09:26:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/113e2326189f4ca69f9b23a66c249d80 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/add36fdabb354104b3566e59eb7dfa37 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db73f996b2614e538611b019be428c60 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cc41215a13240c0abe1208289baed24 : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/753a8417422a4dc99941b40f52d10975 : SUCCESS in 5m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bbfca361f01f4601b7b0f8729c845920 : SUCCESS in 9m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e8725eac58da464cadf9e2123ac50a82 : SUCCESS in 7m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dbadb4159e1c41fabf178a5829f9eb72 : TIMED_OUT in 4h 16m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1f2908a42c6e4de8bdbc5f0a37f8f668 : SUCCESS in 2h 06m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cea06dbfdee7462dbf8c88b2870cf1fb : SUCCESS in 42m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/fcb2392c6ec941d6bd5c640617c2b903 : SUCCESS in 42m 04s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/acf1e3abc74b43dda24a8f1ae6e7e78c : SUCCESS in 43m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1855ed585d044455aa74b7083b673a32 : SUCCESS in 59m 11s","accounts_in_message":[],"_revision_number":1},{"id":"9b13f33d2ed66502625520e1152528e333ad9b7e","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-21 11:05:12.000000000","message":"Patch Set 1:\n\nrecheck\n\nCI Job octavia-v2-dsvm-noop-api  TIMED OUT","accounts_in_message":[],"_revision_number":1},{"id":"33487c63069d845beec7bb8f43a12c86bab6feec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 13:27:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30ca4547bdab4afa857cce2b8bd4d63e : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/425ab3039ec64ca4a7eb0071d12c3d08 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1426c11079b048ef98792a00115d9324 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed85f470008a4c6f8a79080580a78cf4 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c424da33a43e4d9181413f80e9310e58 : SUCCESS in 13m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3df676109af346b48bc2aa2f22693315 : SUCCESS in 5m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f90861da3851491ea2188508c3961d77 : SUCCESS in 6m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cf11467521294f059b24f7cdd1a921dc : SUCCESS in 8m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5cf8c2b1d85c44f2a3ad6682056b5e49 : SUCCESS in 5m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f4731b53967041bd852fa5cb94870f6b : SUCCESS in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a4847fa6299142dd8214be7dbbc24f8d : SUCCESS in 1h 41m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a1de7514f734ea5bc7c609981cc2319 : SUCCESS in 1h 43m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f535f1502ab9400086a0beba8fa1cb38 : SUCCESS in 45m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4cdf79a89fa647168c4692bada8f5adb : SUCCESS in 40m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b4f5b2990490489fb14b98e41f3aa36f : SUCCESS in 43m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aad965813ce04fbdb343da3018186bf5 : SUCCESS in 1h 01m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3cf549e4185a40a79e12f30d317ac8a9 : SUCCESS in 2h 07m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/74b71aadf9e6400f9724abf0758489f3 : FAILURE in 36m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/44c3beb1d7934c119e223c307ab35379 : FAILURE in 2h 13m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/16185ce55a774f5f814ffb9941db06a2 : FAILURE in 2h 20m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a810843e42994dff942f53795dbc79ab : FAILURE in 2h 04m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/078f620620b543a3b21a1bc29d995785 : FAILURE in 31m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"786bda92582a25b6a94747c435c81591e86b49d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 13:28:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b62be4e15279f419451c220debc7d92d53f5cc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 14:22:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/048d0812226c4140be9e27aa77360386 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d12cf43156e474a84f4a408b4de7b78 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6b6eba31d8d426987e6e7c988dbcb37 : SUCCESS in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf6bdb87733346eaaef19b97b9e0822d : SUCCESS in 21m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b2856c408134b58bc8f152befe76794 : SUCCESS in 9m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0a67a0ed5e9f4ad7ace09c4bb4073672 : SUCCESS in 8m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e76ad54701cb42b8a4a0a14085ed5986 : SUCCESS in 8m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7600b0a7445647f28f30b6732aa2cd34 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43e2c608a6ec4f17bf5a943944676afd : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e6eb7e57d44046b99eea39fea581f575 : SUCCESS in 44m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/563ff6a366504f59992781591ed45129 : SUCCESS in 41m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/222c70660ea84871b3fe1901408c5944 : SUCCESS in 41m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/66fd02a1f94a473c960feb6edba7dd7a : FAILURE in 51m 53s","accounts_in_message":[],"_revision_number":1},{"id":"6679837149a4402c873389f500d9776b4d0876a6","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-22 03:10:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"665c240703006fd495381835ecab85ab58cd4347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 05:34:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c50a6fdf9bd64a6da8feb4793f2eb36b : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19c59b5f44ea46c58efcfb841b23abd1 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53c92627462a4920bb509951ebba060f : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9a075f1392a40fa856994e1d75fd818 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09d65fd4556e4e37910ab2f1b233f3a0 : SUCCESS in 14m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/848b00f08ff94a9a8d2e519a59ebffa4 : SUCCESS in 5m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7344d0ceaf1a4cf1b1b2cd01370d0636 : SUCCESS in 8m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/16ad4a1b49bc40939be722b0d1661b70 : SUCCESS in 9m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/89451e1e548e4860802ae71d47eb4155 : SUCCESS in 6m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0dfed7a7dad244e3b352d9c079e278bb : SUCCESS in 7m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cdae7ed2cdad4e3faa8247885d78c646 : SUCCESS in 1h 52m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9b5f0775caa4c85b853c976fb40bbe0 : SUCCESS in 1h 37m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2e775ecb3994454ea4ba6955b399afbe : SUCCESS in 42m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4bcf85f9c4ba44699ce09357c20c9f1d : SUCCESS in 40m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/74852a4c34dd42549124f470f4daa9e0 : SUCCESS in 41m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d6a35284aed34865a7ea583b2aa64e5b : SUCCESS in 1h 03m 37s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/72f874473dab463fb30c3bf609ef6b65 : SUCCESS in 2h 21m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7a1fb31d2d554618a567a25f2807f967 : FAILURE in 36m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/78dd872fa4fb4e5281a8a40b2b6cc1b4 : SUCCESS in 1h 42m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3e2d883dec0841ed89626c1699ac8b7b : FAILURE in 1h 38m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1805fe387ec54fada694e45a2f665897 : FAILURE in 1h 58m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3e2cb9b7e4e94b048b3f53305d514c74 : FAILURE in 26m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24e8445bf93fdf5e30241c6469299a7ba871aabb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 05:34:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27c1095998d20ea746f91a84b6ef80fd07ef736f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 07:30:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1847ac9263d549018a3bbe7aae450547 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44618e2000c34bd18070e867d790fba2 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0062618505649d5b3b77c9022540be8 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e088cd25df94644bd45a78f5143de46 : SUCCESS in 13m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4685a9d6d9f44d289efc1747263610ec : SUCCESS in 7m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/13d995d4d4334013a641553004294bb6 : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cb523cc533074d76839fc1b252edb933 : SUCCESS in 7m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/207007f9527a44568a17737b57bf0442 : SUCCESS in 1h 27m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6a757d44289a46c0b6293755822edf96 : SUCCESS in 1h 53m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ce6e2e0999534817b0b147aa3ecfa583 : SUCCESS in 39m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/80e9bb3b6e3246bba8de88ca1d7c27b6 : SUCCESS in 39m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f9158ae5d884df6bf4f4b0608ead48b : SUCCESS in 41m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b04142d104164ef7983c4ba035048084 : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":1},{"id":"e2c16c6a0913aa66cbd8a4d845c72561b9a4f8ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 07:30:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d6dc6ba870db8e26966c4550434ede572d4aaede","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 07:32:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b100ee365d145fdb3fd32aa2a56b2d9 : SUCCESS in 1m 34s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15beeba8e3104a9b8b71f5b0b0cd2b06 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7dde919290cbdb4d95e0d973046cb75f2e18f12","revisions":{"f7dde919290cbdb4d95e0d973046cb75f2e18f12":{"kind":"REWORK","_number":1,"created":"2021-09-15 03:06:52.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/87/808987/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/808987/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/808987/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/808987/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/808987/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/808987/1"}}},"commit":{"parents":[{"commit":"f3b3e35af34667e2a7d718cc203f2ee03105c29e","subject":"Merge \"Fix race conditions between API and worker DB calls\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f3b3e35af34667e2a7d718cc203f2ee03105c29e"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2021-08-13 06:24:59.000000000","tz":330},"committer":{"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","date":"2021-09-15 03:06:52.000000000","tz":0},"subject":"Fix duplicate SG creation for listener peer port","message":"Fix duplicate SG creation for listener peer port\n\nIn cases where the listener protcol port is same as the peer port\nand allowed_cidr set to 0.0.0.0/0 explicitly, the listener is not\nprovisioned due to duplicate security group creation for peer port\nwith None as remote_ip_prefix. Neutron SG defaults remote_ip_prefix\nto 0.0.0.0/0 if not specified or None and hence the error SG rule\nalready exists.\n\nRemove the duplicate entry from the updated_ports.\n\nStory: #2009117\nChange-Id: I9dbdb71e9b94bbcc75766a8687a996d5358f3381\n(cherry picked from commit 151a94321074b1eda53d1a582fdf26942db85afd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f7dde919290cbdb4d95e0d973046cb75f2e18f12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f7dde919290cbdb4d95e0d973046cb75f2e18f12"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
