)]}'
{"id":"openstack%2Foctavia~808988","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I9dbdb71e9b94bbcc75766a8687a996d5358f3381","project":"openstack/octavia","branch":"stable/ussuri","topic":"story/2009117-stable/ussuri","hashtags":[],"change_id":"I9dbdb71e9b94bbcc75766a8687a996d5358f3381","subject":"Fix duplicate SG creation for listener peer port","status":"MERGED","created":"2021-09-15 03:07:07.000000000","updated":"2021-09-20 17:40:58.000000000","submitted":"2021-09-20 17:38:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808988-story/2009117-stable/ussuri","cherry_pick_of_change":804485,"cherry_pick_of_patch_set":7,"meta_rev_id":"0f11a9deb95049c0cb23c3f2a48387a33c64f90f","_number":808988,"virtual_id_number":808988,"owner":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-20 17:38:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-09-20 14:14:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-09-15 07:03:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-09-20 14:14:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-15 05:12:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-15 07:03:57.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-09-20 14:14:32.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"cf03a367894501a884713391e4dc1c3caa75d17b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-15 03:07:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"297c1efdc99562ba77850c1dc4ba5c918ec3cf61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 05:12:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72853084233a4445a9625137a2423e7e : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21dcd6f1b1c54f5e871814b685821bfd : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bbf5bbc8031497484ef3228897bf1d5 : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/efd1f7ed159247dda86050e79e146f6c : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b60b982f8b345a4ab0d5c5500f64e38 : SUCCESS in 12m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a306f93d00fe44bfa10f6a3ab7cfd37b : SUCCESS in 5m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/34828d330cfa41029bf09b27681d9a85 : SUCCESS in 5m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3c8efaef9af747089846e3495c42cb55 : SUCCESS in 8m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fcf5a1b462404fb0a42eb1791e4c21cb : SUCCESS in 7m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5ad8470227a44f45b464ce2a040cd423 : SUCCESS in 1h 02m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c48bdda4cdea4335b7a718010f4ee02e : SUCCESS in 1h 29m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/14a711dd77b340cdbe6f31a0292b1ff0 : SUCCESS in 37m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3e34eb433b9d4e6aaf366c433dfc6ab3 : SUCCESS in 33m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d19ddfb0c0944ce7a7c5c96a7b612f3b : SUCCESS in 38m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dfd007b7d7764ead96b9cba7b0f16eb9 : SUCCESS in 51m 16s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d501f7901e454caaa97cff78e5484dad : SUCCESS in 1h 30m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/887a5d6578034ebfa8054fb67d497459 : RETRY_LIMIT in 20m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b2125e8b2c0342c796cf7c5dd80a2528 : SUCCESS in 1h 27m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3adf7b1a91c9d5f31d6ce04f95e7c85c7a4af47c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-15 07:03:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"03f103e7c5e2d1ed8545c059b44a33dabbd2d772","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-09-20 14:14:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"15ed2ee8a3d79f406845c774da4dbc2adebf1676","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 14:15:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4140e8f3b1346fa818b23d4d53c85383187dfe8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 17:38:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53bd78f81774483e92580dfb010c24e2 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/726e1bbc2ea94417a9085bc7b036fe4e : SUCCESS in 10m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0468e7d0f72401e830171a2e18f2090 : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2af65e31dd1b49b0a43925766ff13d91 : SUCCESS in 12m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f12a79112bb945219a89294c529c3f6e : SUCCESS in 8m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8d86182809924a559b7fede6ebbacdb1 : SUCCESS in 8m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/95617d7932704fc38b00c51f136da6e0 : SUCCESS in 1h 16m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0c793863d5944159fab9976c55bcd77 : SUCCESS in 1h 57m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1715b314f7bd4a919b882b44601b5520 : SUCCESS in 38m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/45cf7e3367174258a1d1f07aee609183 : SUCCESS in 37m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a60abf9bf4c8401a85401085347921a4 : SUCCESS in 38m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/28f42daec52d4dfc8d41379d5cc5adf3 : SUCCESS in 1h 05m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ec912cd3b5d7ae602cb1d55ef9360735511b93e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 17:38:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0f11a9deb95049c0cb23c3f2a48387a33c64f90f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 17:40:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f517f938ea4145bf9d58f22e67785157 : SUCCESS in 1m 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6562469da3948fe85893a646fd2d0d9 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bfdc3614e0383f3d7b46b178f733cfeced6642c5","revisions":{"bfdc3614e0383f3d7b46b178f733cfeced6642c5":{"kind":"REWORK","_number":1,"created":"2021-09-15 03:07:07.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/88/808988/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/808988/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/808988/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/808988/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/808988/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/808988/1"}}},"commit":{"parents":[{"commit":"d79c80c39c6953768c267abdba0c38a7374d6e2b","subject":"Fix race conditions between API and worker DB calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d79c80c39c6953768c267abdba0c38a7374d6e2b"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2021-08-13 06:24:59.000000000","tz":330},"committer":{"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","date":"2021-09-15 03:07:07.000000000","tz":0},"subject":"Fix duplicate SG creation for listener peer port","message":"Fix duplicate SG creation for listener peer port\n\nIn cases where the listener protcol port is same as the peer port\nand allowed_cidr set to 0.0.0.0/0 explicitly, the listener is not\nprovisioned due to duplicate security group creation for peer port\nwith None as remote_ip_prefix. Neutron SG defaults remote_ip_prefix\nto 0.0.0.0/0 if not specified or None and hence the error SG rule\nalready exists.\n\nRemove the duplicate entry from the updated_ports.\n\nStory: #2009117\nChange-Id: I9dbdb71e9b94bbcc75766a8687a996d5358f3381\n(cherry picked from commit 151a94321074b1eda53d1a582fdf26942db85afd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bfdc3614e0383f3d7b46b178f733cfeced6642c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bfdc3614e0383f3d7b46b178f733cfeced6642c5"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
