)]}'
{"id":"openstack%2Foctavia~809122","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I990b6af1ff880b25e54f6c41ae0c966007f3f098","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I990b6af1ff880b25e54f6c41ae0c966007f3f098","subject":"Fix MAX_TIMEOUT value for listener","status":"MERGED","created":"2021-09-15 06:41:41.000000000","updated":"2021-10-06 21:56:59.000000000","submitted":"2021-10-06 21:54:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809122","meta_rev_id":"d089c2278aa400ab09ad340a09ca4d223f1c027b","_number":809122,"virtual_id_number":809122,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-06 21:54:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-09-20 14:14:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-09-21 10:41:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},{"value":2,"date":"2021-10-06 20:19:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},{"value":1,"date":"2021-10-06 20:19:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-15 06:41:41.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"state":"REVIEWER"},{"updated":"2021-09-15 08:27:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-15 18:02:17.000000000","updated_by":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-09-16 19:58:22.000000000","updated_by":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2021-09-16 19:58:22.000000000","updated_by":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-09-20 14:14:17.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-09-22 10:49:39.000000000","updated_by":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"}],"messages":[{"id":"a900e0439af75d2741e51a4d7c79f16d5a747d37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-15 06:41:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ee23fc3cd26970b2ea731574c75756524de09b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 08:27:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e15849d23e8845a595393c0e90cf5956 : SUCCESS in 24m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04c8cbb5185847b1b6fe96d8fd2186e5 : SUCCESS in 21m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0469c585172a47a0b13be553020859cf : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fb7e5e40aaa4cd7b1abe4403cbdfffd : SUCCESS in 20m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c58fa557e0374eb381241dd571a567a1 : SUCCESS in 18m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d4952edc08f4d0484a4c5ec5c9f7331 : SUCCESS in 12m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e46a4ed4e7d4be88ef640e5f9b116dd : SUCCESS in 6m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a4df7394cd8a407f99de2ce9697fae0d : SUCCESS in 21m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1c07e78656ab4c64976aac5f3bb20b2f : SUCCESS in 25m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f8b0e5cd18264cc99ebab215dd61a053 : SUCCESS in 7m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0a1078997a954ce9a4f135cb6643436b : SUCCESS in 19m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee6219a2501d4f54bc55d0d01f313cbd : SUCCESS in 1h 13m 41s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/455932f747944e378883d678d44cb09a : SUCCESS in 1h 12m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d358dcd12d9c428585e74a6b7d5f9bba : SUCCESS in 1h 25m 01s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ee19557c5e3a48a0ba69e871ab389751 : SUCCESS in 1h 43m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fd7bf3bc044f49c7ac08146460c16340 : FAILURE in 12m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/94e06267b9bf4041a36b6d9191482610 : RETRY_LIMIT in 11m 57s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/49832c6431134051a461a217fd80e171 : RETRY_LIMIT in 11m 17s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fd72a8f0984c4bcf98a968d30b0ed6b9 : SUCCESS in 37m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6b09b9c72025451ca381d898fccfcc49 : SUCCESS in 1h 28m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a543da3e3757434d8f8758885f84bc20 : FAILURE in 35m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e109b9f32233472588cce8e2974c497d : FAILURE in 35m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e0f72ab5b36744c08feaf3cfc40645d6 : FAILURE in 12m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6ce3d09a2fb7853102538ed0ac505bedfc074ba","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-09-20 14:14:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2a6fa1c114a893df5a5afe4483bfd91390ce8209","author":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"date":"2021-09-21 10:41:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bdc0ca380b18c351b4add4df94fda1123b2413f6","author":{"_account_id":31754,"name":"Roman Goncharov","email":"gadzhet007@gmail.com","username":"gadzhet"},"date":"2021-10-06 19:17:03.000000000","message":"Patch Set 1:\n\nHey, core team!\nPut your regal wf+1 and let\u0027s close the last cherry-pick!","accounts_in_message":[],"_revision_number":1},{"id":"0a94a790056672dcbff806090ca55ee63b609aa1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-06 20:19:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"0069334fdf68179d1b7d338109b3f09c77fbd852","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 20:20:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"093194de251238282a29e4c44b7c3c91905e191e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 21:54:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e3334722fef45e3b88fad0b4a3e3f9c : SUCCESS in 17m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f39dbd00ccd6424e89298ee2535fc8f8 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e9701170f204c9fbd6a4fa53eedae32 : SUCCESS in 17m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b38a708be6b34f64a4d64812db965880 : SUCCESS in 21m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a10525bf1fcb416e97d8794202087311 : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3f69e730e5b4457a5d814808a74386e : SUCCESS in 5m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f42f7bd8f38e454fbf21ef59e7cd2f28 : SUCCESS in 7m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/93079158e0914b5d89f9329abe41f782 : SUCCESS in 20m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/31ccc31802f24a38a3ccd4074de7918c : SUCCESS in 1h 05m 59s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/05e41a3b498b4704a650b785ed149823 : SUCCESS in 1h 08m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e31e73e05e10439592c1b9f0c3c9beac : SUCCESS in 1h 30m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b1c1e294c9dd4a8e2d62cbb0684ca1b61c5a7eb1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 21:54:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d089c2278aa400ab09ad340a09ca4d223f1c027b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 21:56:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a89f23d733849eaa2ffbbe024858ca9 : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f556dc06ca2b483b9f527806bff36ef2 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"604cc30bd3b19b1466a14d4f180986188a7d512f","revisions":{"604cc30bd3b19b1466a14d4f180986188a7d512f":{"kind":"REWORK","_number":1,"created":"2021-09-15 06:41:41.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/22/809122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/809122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/809122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/809122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/809122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/809122/1"}}},"commit":{"parents":[{"commit":"b1cc4467a916fea026454a7c3635a3681ab34253","subject":"Enable lo interface in the amphora-haproxy netns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b1cc4467a916fea026454a7c3635a3681ab34253"}]}],"author":{"name":"Roman Goncharov","email":"gadzhet007@gmail.com","date":"2021-09-09 15:04:12.000000000","tz":180},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-15 06:41:20.000000000","tz":120},"subject":"Fix MAX_TIMEOUT value for listener","message":"Fix MAX_TIMEOUT value for listener\n\nIn the database, timeout_client_data, timeout_member_connect,\ntimeout_member_data, timeout_tcp_inspect have an integer type with\na maximum size of 2147483647. Now MAX_TIMEOUT in API does not exceed\nthis value and is equal to 24 days (2073600000 seconds).\n\nStory: 2009193\nTask: 43248\nChange-Id: I990b6af1ff880b25e54f6c41ae0c966007f3f098\n(cherry picked from commit 3a8f056306bba4502639bfc89752d8a48d694e7f)\n(cherry picked from commit 9a25a643e2cb544bd9b4993a126a74ecc482ac2d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/604cc30bd3b19b1466a14d4f180986188a7d512f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/604cc30bd3b19b1466a14d4f180986188a7d512f"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
