)]}'
{"id":"openstack%2Foctavia~809287","triplet_id":"openstack%2Foctavia~master~I1aab9faa8afe845e445e620d1800785d2e19ad1e","project":"openstack/octavia","branch":"master","topic":"fips-compatibility","attention_set":{},"removed_from_attention_set":{"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2022-01-19 23:53:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1aab9faa8afe845e445e620d1800785d2e19ad1e","subject":"Update scripts to use fips allowed algorithms","status":"MERGED","created":"2021-09-15 20:57:42.000000000","updated":"2022-01-19 23:56:33.000000000","submitted":"2022-01-19 23:53:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809287-fips-compatibility","meta_rev_id":"2a17abf36fc8d1f21add7ab7c17b82612791db56","_number":809287,"virtual_id_number":809287,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-19 23:53:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2022-01-19 23:56:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2022-01-19 23:56:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2022-01-19 20:33:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}],"reviewers":{"CC":[{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"}],"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-16 01:04:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-17 14:37:01.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-09-17 14:37:01.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-11-17 19:25:23.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"CC"},{"updated":"2021-11-18 11:02:25.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"e5b640fb66fe9028075205b1f68c189cf28b7f14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-09-15 20:57:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9909e0acbc527ba63c9cd5152472381d6e22c5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 01:04:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fee727126b647b29ee472c6a9589eec : SUCCESS in 12m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04d2659203364b71afed4fe7830fc828 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bf3ff3d2ee2427f9d58633994d9d6c7 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe26854be52343208036ab0e0d2bee90 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9068b8b8387434bb72513160d5cf897 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36fb4487c92643959cb22da0021e0ff4 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a8c2d401a2f45bbbaed5027603b22d5 : SUCCESS in 13m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e8017d6d76b64adeb4625115a80450e9 : SUCCESS in 5m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5040f92cc06b4c63b25c000f6694ff4c : SUCCESS in 9m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b282c217526e4625a540de5ac041985d : SUCCESS in 5m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3b2396f47b534b9d99550ecb8dc840d0 : SUCCESS in 9m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9647101f920847e088d1087939d03e69 : SUCCESS in 2h 14m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/471df9b6cbf94298b09eb598cf92b6fe : SUCCESS in 2h 06m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8ba0c097ec7943529659e7b97658d6cb : SUCCESS in 57m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7702da4cf8a4408eb8ab138b0b1b29ec : SUCCESS in 1h 12m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f6e52e2518924d439b30a7055af05f70 : SUCCESS in 35m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4cfddebe636f4288b49d5b197fe233cf : SUCCESS in 1h 40m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d53e64fa453c4ea5ad4012bbaf7bad2e : SUCCESS in 2h 15m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7de7ad107b934c3082adef6e38055b29 : SUCCESS in 1h 55m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e388adbb77be4cbcb25187cdabe999ef : SUCCESS in 2h 02m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4120046c1a4445e7abb44d41d3482782 : SUCCESS in 1h 56m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/be26e0b5c6114f7bbdc4c27d7531404d : FAILURE in 9m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9ca093fb647cd6edb80add5a6c2f560fe9f6a043","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-10-15 17:22:01.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"6f819722d7f1c3d421c1d9e0aa47e929e144b212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 19:48:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c16494ab0aa41e09668ab42a04e0e9a : SUCCESS in 15m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c5622181c8b4e76871e7937ecd9a03e : SUCCESS in 12m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a78fce378079401da990a6c991de18da : SUCCESS in 12m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1eef558efb9450f86ad9f40200261ca : SUCCESS in 17m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d16041069ecc403facfae89e4151028d : SUCCESS in 12m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a47163205f94010ba50702609b58b2f : SUCCESS in 12m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ef4cf9bf4ab4dc38af847e280f3a49c : SUCCESS in 14m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c8497924b4f24764924c47d8dace0a00 : SUCCESS in 13m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b6b82529c6c743c9be9741dae82c8d33 : SUCCESS in 16m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d31c813317ed44b18862aa349f38aa8a : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c51c2506d0ad4c3882b524a79117acd6 : SUCCESS in 17m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/48d531b3b3344d64a7d7a6c90d58fadb : SUCCESS in 2h 13m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b062277db1824525863596af54cbf7a8 : SUCCESS in 1h 40m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/52aec9558c4b494b95b404cdd284d5c9 : SUCCESS in 48m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e3c20b4e8cc74fdf94860d9b848278c2 : SUCCESS in 1h 01m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8b9343a40fe5437d9dcb52d028d3018f : SUCCESS in 46m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5ce3c5cf6fec4f0caf94f4d35c987240 : SUCCESS in 1h 25m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1c34e06eee7f43a9b0baa85f784ebe4b : FAILURE in 2h 23m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/70bc9f2b52304cb09ebeab38b8157b08 : FAILURE in 1h 53m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7497b716ce664138aa701f4c2071bca8 : SUCCESS in 1h 25m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8f35d5e1c68c42bcbcce94e42cd848c7 : SUCCESS in 1h 21m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c2ff286908e74ccb870596a426850539 : FAILURE in 11m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"34179fcde433cd2885470bfbb4065305a8d795a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-11-05 20:39:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"8733aead4e74c740e3d2451e0e6e6bf136910c31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 23:00:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8abe323517ab4e2a9ec2083591f371f2 : SUCCESS in 11m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c0568d533a84276bf6fe98527273d8b : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7ab033871db4cad83c058fa72f4f62b : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fcda8da46a54e378da888ab82f5ef55 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da183e0eb6be482a87a539c51cdbdf4f : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a2e807afdda4175be0516c3729c4aa3 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffd798ceb02b4817921c38c1c8394622 : SUCCESS in 13m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cfbcf2b164f44639b7c3898fbe71d975 : SUCCESS in 6m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/17431ccf38b1430a933b33a6179cebf5 : SUCCESS in 9m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/43f71078b23145a7aa12910fc216f277 : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/94a00e966bd04cfc979b20dd043d130a : SUCCESS in 8m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0c8262616ccd438091e698c3ebe3a0ed : SUCCESS in 2h 13m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc42e22803c5482da20ce81e312180f8 : SUCCESS in 1h 26m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dd85ba257bf7436b8675fa09e42a2297 : SUCCESS in 40m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/095a44d8f1a84e509cd47be103244cff : SUCCESS in 51m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f2f410715d64baf99a1255c0545c102 : SUCCESS in 35m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b988c2cc68394375a06bfa5b686a1dac : SUCCESS in 1h 35m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d6bcb7793d1b4c7291e383f663a0064e : SUCCESS in 1h 46m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0b72179c3c3e4360a45dd71df7981eda : SUCCESS in 1h 31m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/98b2058b9be548158a5eea8a39a74c6a : SUCCESS in 1h 36m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/19b8d044033344d7b96720349e3ced45 : SUCCESS in 1h 34m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9305dd099daa47e9886c5b7f410e30d3 : FAILURE in 24m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7915bc5aff95343092ef05551b8766dc2ef63c8f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-11-17 20:52:34.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3e0fa19c5be392fa99e59607f2ce372c5fa17a73","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-11-18 11:02:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"918e3599d874d1d8c256b72d2989fc10457d65ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 11:03:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"63490224d51ffaaa7f4b8db55cd77da24c37aaa2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 11:44:50.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db656ffdadd6497fb3f1a7ee8e49b3dd : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08ddb5ab25d147e696b93559d470131d : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/143e1354781a4dfe9cfb5dc46195d26c : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2332e4bbd46e4d7bbfcb41e4b2f844be : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a6ea8db8764a168d1c9d27507d66c8 : SUCCESS in 11m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6c77df821b084ef782d3709b2d8d411f : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/641c8826be8a44e0a7a41d51cc1cdef4 : SUCCESS in 21m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/925e773e2f6a4f3b8d3cc398f6dc3b5b : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c6f32931935403db25d924b76fcba5d : FAILURE in 31m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cbe6385bab314a34894a16126ad76076 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f474ad6885cc435b920934b70af35e27 : CANCELED","accounts_in_message":[],"_revision_number":3},{"id":"dbe30c03af74cbb66e85b41fe0c7ec7dc5015e40","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-11-18 15:39:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c9f7341bbac0f7668f083347941a436ecb7d737c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 18:02:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7206e93eca3b42dba6915bd5ff8f390c : TIMED_OUT in 52m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a42fc1d28e844e79b46557bf94ffd318 : SUCCESS in 30m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14b66f69319041dc8008704a246d940d : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cd39a60d8654632bb45e7f6ac4b49be : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d23f903cc8a4d80bd232a904fa31869 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/deb31fc4284f4d0daeadd3a5df658d63 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6579054fa6e4138ba56691c93d8223f : SUCCESS in 14m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e758e730ed3945f38100b6fb00f9a8fb : SUCCESS in 6m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/425abb715bc94132995cdbb18f7d54de : SUCCESS in 31m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8d1d7d05b2934cf5bf8c9532a6f077ae : SUCCESS in 6m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9814ef08becb4d21b66d7964a499a3f6 : SUCCESS in 31m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/230b76d3b0f7415287e4ce18abc16727 : SUCCESS in 2h 13m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/488fa42e40844d2582a65b08ea4d3606 : FAILURE in 31m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c3f2d454dd0d44cf9c9dea5ebba26c3a : FAILURE in 31m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/453713724f1f44fe95d75e85ee633efd : SUCCESS in 49m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/416a9905da7942568393762cb5b286e9 : FAILURE in 33m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8ce6098bc76e422bbd3964141ca28856 : FAILURE in 35m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9df5bc9f98cb4270b936db9ee7987a3e : FAILURE in 42m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bc324147562043528fefd0e3bf108fb1 : FAILURE in 32m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7a52e1d190024ca6a1fd4221223e3402 : FAILURE in 32m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/afe54d6e91e1464d92a293d65efbfe69 : FAILURE in 33m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/eb8700716f9d4b35b068763930fd2bb1 : FAILURE in 10m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"76d94e4fabad3bacb5cdb1d9a622fe601daa66a9","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-01-17 18:53:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a239d3b12dfde3adf8e2f62fe690a615b1396ef8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-01-17 18:54:09.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"e9d95cd5d2635340c1a32eca063fa81acba8f3cc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-01-17 18:54:30.000000000","message":"Topic add_fips_job removed","accounts_in_message":[],"_revision_number":4},{"id":"d1839125451328c5f2b930ac387e7b12511ed514","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-01-17 18:54:39.000000000","message":"Topic set to fips-compatibility","accounts_in_message":[],"_revision_number":4},{"id":"e8db33cd50bf72e54b3af124c1ab68ddd0e0e88e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-01-17 19:03:22.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent, and commit message as Patch Set 4.","accounts_in_message":[],"_revision_number":5},{"id":"9635df5096f7ba9c848864a2f4efedaf139f006b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 21:29:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d82628888392499ea745718b90d57895 : SUCCESS in 10m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5bc77895ba9f4e34aa8d26647de64afb : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53a95e37c67948fd9aaee010a095d7e3 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13655aef96134ee0bcba2e6a92d08165 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36ee4e0b1b0c4b14a2438c2a51ad1010 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12326e5601f84a3db58932c18c81710c : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8081214be37347058ca1447e6af13587 : SUCCESS in 10m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/12619b03a3ee4bd3b3dbdf3e420b1734 : SUCCESS in 6m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c668e1fc78d64e91a8abef9a7af1c7f0 : SUCCESS in 8m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9fc7bcd4f79a47ae8d5719ffcb1e698a : SUCCESS in 6m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5c2a36143ce241fcb8bf8d304cacde9d : SUCCESS in 9m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/03e0f08fae324ae6aa2ef8ac7f197446 : SUCCESS in 1h 42m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c4355410b4b04570a82c91df6bd8a627 : SUCCESS in 1h 29m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/99e70f001a404d6dbaaa5f8ffee031f9 : SUCCESS in 33m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3b5fbcd0eea141fa80b16f9f94e6dc3f : SUCCESS in 45m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/477a732bf7db48a68f0f1e968183bbdc : SUCCESS in 45m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d810ed6ae6c74925b5cd483e0037cec2 : SUCCESS in 1h 57m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/58f478487f1645ab932fa0f1f1ad6754 : SUCCESS in 1h 59m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/82ae56162ea74ef3b219f6e4e68fc392 : SUCCESS in 2h 09m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1cecbdecc29440fa8c38a50225fa9114 : SUCCESS in 2h 22m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/943e007bf95b46b6b84b73b35e2008e3 : SUCCESS in 1h 46m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd128f38726a479ca36270043772da39 : SUCCESS in 2h 00m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c8a284f690f63134193876203e6d95f37aea8f94","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-18 00:32:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"96fd5931e172fa1163f69e45c863082d943a6531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 03:02:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ef3f4c04bbb41b09451b69e74b0f8d0 : SUCCESS in 9m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c5fadcce43b41f3b2bd233c219188a5 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f96e4fac02af488697008a460c080557 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/987d72590c6440eebab95be2a8c5ce01 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1458ef56aef2450dbe15b8cfa8cb077d : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06cfa1f56df24480a19a0ef02a5e08fb : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9361941cfae142f78f0da0b619cb950a : SUCCESS in 13m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f4e61f595b4e47b49305aa2c99f3b8be : SUCCESS in 5m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/580f030b1b41432da98a0fdbb3575cdb : SUCCESS in 9m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3a71732a2c5e4e328d079c664177c6fd : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5583e637e79f40b09198e990c8e5827e : SUCCESS in 8m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0fe4efa2dc774a8193288e78c1b7a231 : SUCCESS in 2h 10m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/91ceb4067ea14d3095d8bb17a13c779b : SUCCESS in 1h 48m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bc3e9ac6f8c3442ab189a5ea87e17c74 : SUCCESS in 34m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5312f7e8466d43a5801f094905b142e5 : SUCCESS in 53m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/156bb30140e94b8ab3a0203e0d5d6e0f : SUCCESS in 32m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7057c5be021341b2998cea06e418e09e : SUCCESS in 1h 37m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d5b2d442bd2d417bbaa7d2c77010811d : SUCCESS in 2h 18m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/de5d109a351940bcbde9066dc73a5037 : SUCCESS in 1h 58m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5518457e6ee64e418dc61e094a5ad70c : SUCCESS in 2h 17m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e30da0c2684a419aad68f613f460d58f : SUCCESS in 1h 40m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/690361977bd04180a4efcb3f9c1b4d2c : SUCCESS in 1h 45m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ab554d1fc2884570754d4056c50eefe5e090292c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-19 20:33:02.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a684aaedb7a1560644a924e0207a80d347df4297","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 20:47:55.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8a6dbc68d76974c03f9eb341b9886f20b63e5d63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 23:53:39.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f022b45d8e2f428fbbe08a0050d5458d : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fac8ab0405a4be5884a1b33698476d3 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45b6e2df7afc4268a497e57fcc4737e3 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f06d4f0116c34c419d21adf9fc114737 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e860a1837a294f55a25c69ba5c207f37 : SUCCESS in 12m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c109a9f395de4df3a85f24a8ac5a864c : SUCCESS in 5m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7d492091d99b436b89c8715d6d9537f9 : SUCCESS in 8m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7a6cc8dfc65c41f3a783bddac75d4a25 : SUCCESS in 2h 20m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/35f47e81b0384a00a37f64104b23b822 : SUCCESS in 1h 40m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9f38a5c04e4c4effb03101265d5eac36 : SUCCESS in 39m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d1be5591d88f447599ee9a54320700ad : SUCCESS in 1h 01m 52s","accounts_in_message":[],"_revision_number":5},{"id":"ceffde173833fef7055c347e1f4d8c51f2ca8618","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 23:53:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"236d15942649c13762378a39165156731031db00","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 23:56:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d8488b9b65a40d385d4d6ae372aa8a1 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"36a642d9d0b95b6337558144450bbc7802784c23","revisions":{"40055275f1ea67657a3b01854e1aa467956cce99":{"kind":"REWORK","_number":1,"created":"2021-09-15 20:57:42.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/87/809287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/809287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/809287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/809287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/809287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/809287/1"}}},"commit":{"parents":[{"commit":"b8ddbc94a1ff1c8dcb3414afd1d5f5269a09f9da","subject":"Merge \"Disable conntrack for TCP flows in the amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8ddbc94a1ff1c8dcb3414afd1d5f5269a09f9da"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-15 20:51:07.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-15 20:51:07.000000000","tz":-240},"subject":"Update scripts to use fips allowed algorithms","message":"Update scripts to use fips allowed algorithms\n\nOpenssl genrsa is deprecated in favor of genpkey, and fails in FIPS mode.\nUpdate the relevant calls to use genpkey instead.\n\nChange-Id: I1aab9faa8afe845e445e620d1800785d2e19ad1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40055275f1ea67657a3b01854e1aa467956cce99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40055275f1ea67657a3b01854e1aa467956cce99"}]},"branch":"refs/heads/master"},"ff8cd640a48708dc88511f1660788f966b234cf6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-15 17:22:01.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/87/809287/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/809287/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/809287/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/809287/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/809287/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/809287/2"}}},"commit":{"parents":[{"commit":"0b84631948bd3c5eb25a0921d5fbad657036304f","subject":"Merge \"Increase TCP buffer maximum and MTU discovery\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b84631948bd3c5eb25a0921d5fbad657036304f"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-15 20:51:07.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-10-15 17:19:52.000000000","tz":-240},"subject":"Update scripts to use fips allowed algorithms","message":"Update scripts to use fips allowed algorithms\n\nOpenssl genrsa is deprecated in favor of genpkey, and fails in FIPS mode.\nUpdate the relevant calls to use genpkey instead.\n\nChange-Id: I1aab9faa8afe845e445e620d1800785d2e19ad1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff8cd640a48708dc88511f1660788f966b234cf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff8cd640a48708dc88511f1660788f966b234cf6"}]},"branch":"refs/heads/master"},"a32312167c56f26251ec9926e53081c388f89d88":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-11-05 20:39:25.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/87/809287/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/809287/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/809287/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/809287/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/809287/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/809287/3"}}},"commit":{"parents":[{"commit":"81134bdfec79c8019c26e7846d31a2b2952965a0","subject":"Fix failover of az-specific loadbalancers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81134bdfec79c8019c26e7846d31a2b2952965a0"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-15 20:51:07.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-11-05 20:39:25.000000000","tz":0},"subject":"Update scripts to use fips allowed algorithms","message":"Update scripts to use fips allowed algorithms\n\nOpenssl genrsa is deprecated in favor of genpkey, and fails in FIPS mode.\nUpdate the relevant calls to use genpkey instead.\n\nChange-Id: I1aab9faa8afe845e445e620d1800785d2e19ad1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a32312167c56f26251ec9926e53081c388f89d88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a32312167c56f26251ec9926e53081c388f89d88"}]},"branch":"refs/heads/master","description":"Rebase"},"09ac2ec8befd51e118423462abc29521536500b6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-01-17 18:54:09.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/87/809287/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/809287/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/809287/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/809287/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/809287/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/809287/4"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-15 20:51:07.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-01-17 18:54:09.000000000","tz":0},"subject":"Update scripts to use fips allowed algorithms","message":"Update scripts to use fips allowed algorithms\n\nOpenssl genrsa is deprecated in favor of genpkey, and fails in FIPS mode.\nUpdate the relevant calls to use genpkey instead.\n\nChange-Id: I1aab9faa8afe845e445e620d1800785d2e19ad1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09ac2ec8befd51e118423462abc29521536500b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09ac2ec8befd51e118423462abc29521536500b6"}]},"branch":"refs/heads/master","description":"Rebase"},"36a642d9d0b95b6337558144450bbc7802784c23":{"kind":"NO_CHANGE","_number":5,"created":"2022-01-17 19:03:22.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/87/809287/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/809287/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/809287/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/809287/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/809287/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/809287/5"}}},"commit":{"parents":[{"commit":"27e5b27d31c695ba72fb6750de2bdafd76e0d7d9","subject":"Merge \"Switch centos8 jobs to centos-8-stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27e5b27d31c695ba72fb6750de2bdafd76e0d7d9"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-15 20:51:07.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-01-17 19:01:16.000000000","tz":-300},"subject":"Update scripts to use fips allowed algorithms","message":"Update scripts to use fips allowed algorithms\n\nOpenssl genrsa is deprecated in favor of genpkey, and fails in FIPS mode.\nUpdate the relevant calls to use genpkey instead.\n\nChange-Id: I1aab9faa8afe845e445e620d1800785d2e19ad1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/36a642d9d0b95b6337558144450bbc7802784c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/36a642d9d0b95b6337558144450bbc7802784c23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
