)]}'
{"id":"openstack%2Foctavia~809522","triplet_id":"openstack%2Foctavia~master~I193abaebdaca10d353e2dbaa14df110048ffaa58","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I193abaebdaca10d353e2dbaa14df110048ffaa58","subject":"Fix some new pylint errors","status":"MERGED","created":"2021-09-17 04:59:18.000000000","updated":"2021-09-20 17:40:48.000000000","submitted":"2021-09-20 17:38:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"809522","meta_rev_id":"91c39a63d0f72dfff43ec384723b0dad03bbdc33","_number":809522,"virtual_id_number":809522,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-20 17:38:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-09-20 14:13:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-09-17 15:14:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-09-20 14:13:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 07:37:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-17 15:14:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-09-20 14:13:36.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"d6dd114a5c77d2e3462b4d59a08cbd6775ef0255","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-09-17 04:59:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87b974440139b6127b622e2a20453a0b606ae759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 07:37:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6ba5939f95243be8bc7d72400620f92 : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34c3b70738b44cc19c4b5f5be96417c7 : SUCCESS in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71e2db81a78346c6bd7423f01b53616d : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/927fd0cd7c6f49bfafd574aa12fc6a60 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78a6eb5f18e846c9a21bfe98d97ca962 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1360c595e2fb49628486c39eb23c31af : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42a4fe00730a49999fd6f0a0eb6993f7 : SUCCESS in 13m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/84a1a86d31fa4f9cb79c7c00e539d8f8 : SUCCESS in 6m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/926e273be62645bb8707eee69a744b8e : SUCCESS in 8m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9cbb9c44adf14c0ab00fc41f79a57e5d : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/54348230c79b47fc961db5e4d7b380e2 : SUCCESS in 10m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2a848522ac4f4ed6b1e01ffde2b9f4f7 : SUCCESS in 2h 12m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a08862c3797943389570679aab886b92 : SUCCESS in 1h 56m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0f2a6a57901a40cd8b5e30db3bf2d13b : SUCCESS in 39m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7fa581c5f0a545d7b1142f9600ef4520 : SUCCESS in 1h 01m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54e223878355400bae734cce8e1d3a15 : SUCCESS in 47m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/281ed964807b43438519e8bd79b2bc3a : FAILURE in 1h 50m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2b8e76349743436abe92a0d7a6cea0eb : SUCCESS in 2h 33m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/78a800561c1d4c66842bcdef7dcfae87 : FAILURE in 2h 19m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6242c9e22b2244bebdc1bab982aec1e7 : SUCCESS in 2h 03m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/093049cbfa1b40f788ec340b90be50d5 : SUCCESS in 2h 12m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1f60fdec1a03490998fd9e695f8d391c : FAILURE in 13m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f96f8f710631b80b49711d773121c885c058de3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-09-17 15:14:49.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"aa55aed971c9ba90b2e259f313026803879e8c59","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-09-17 18:13:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b2f916e60038a9e7fccc1d7d45e733bb8c1092e8","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-09-17 18:18:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c14a4e836637399e364b4ebee3cfae005b9995c6","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-09-20 14:13:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d13372924c49a74e98884ff108b69d91bd42ed64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 14:14:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6cdf4f62fa7e2e22869a5a4b140c2d1e14b492af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 17:38:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a57c6f6d801f4a29bde76e58b3227fb0 : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd3fadf3fe7d4c4c87a26cc966202993 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5eb0ef10c0c640609c556aa306239a26 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dca3a529217647acb91223d6e92e0d18 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4262ead576b6481ca19d6ea9db974ad0 : SUCCESS in 12m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c204cc8395b3480b881a642d9c469e5e : SUCCESS in 6m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0287db6f099f4737ac2e54a0d49734de : SUCCESS in 9m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5febbb0b84a3408ba8a12f610e781f0f : SUCCESS in 2h 25m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/976f56101a1e4415a894608153425722 : SUCCESS in 2h 02m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b863b16e5e88485a83ffdd10c0877669 : SUCCESS in 50m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b3b125c061c54c0fa89011d247ff3a04 : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d39ff7f082f23556fbfcc1f5655dd6d8e4b9e0b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 17:38:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"91c39a63d0f72dfff43ec384723b0dad03bbdc33","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 17:40:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d05d79e0cd394e1d99128e028942b725 : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba5b4a7fb514d1e8c9cf03c03c356aad31acdf6a","revisions":{"ba5b4a7fb514d1e8c9cf03c03c356aad31acdf6a":{"kind":"REWORK","_number":1,"created":"2021-09-17 04:59:18.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/22/809522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/809522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/809522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/809522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/809522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/809522/1"}}},"commit":{"parents":[{"commit":"b8ddbc94a1ff1c8dcb3414afd1d5f5269a09f9da","subject":"Merge \"Disable conntrack for TCP flows in the amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8ddbc94a1ff1c8dcb3414afd1d5f5269a09f9da"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2021-09-17 04:46:37.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2021-09-17 04:58:43.000000000","tz":-420},"subject":"Fix some new pylint errors","message":"Fix some new pylint errors\n\nAlso squelch warnings about f-string usage for now (unless folks think\nwe should do that entire refactor here).\n\nChange-Id: I193abaebdaca10d353e2dbaa14df110048ffaa58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba5b4a7fb514d1e8c9cf03c03c356aad31acdf6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba5b4a7fb514d1e8c9cf03c03c356aad31acdf6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
