)]}'
{"id":"openstack%2Foctavia~809801","triplet_id":"openstack%2Foctavia~stable%2Fxena~I265b98520fe9708d908079a9dd540b9d6b2de68d","project":"openstack/octavia","branch":"stable/xena","topic":"create-xena","hashtags":[],"change_id":"I265b98520fe9708d908079a9dd540b9d6b2de68d","subject":"Update .gitreview for stable/xena","status":"MERGED","created":"2021-09-17 16:11:43.000000000","updated":"2021-10-06 02:55:53.000000000","submitted":"2021-10-06 02:53:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809801-create-xena","meta_rev_id":"00915d97e58cde2253ee109b86f1fb07429be64b","_number":809801,"virtual_id_number":809801,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-06 02:53:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2021-09-24 12:00:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2021-10-05 15:56:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2021-10-05 18:35:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 16:56:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-24 12:00:20.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2021-09-24 12:00:31.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-10-05 15:56:05.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"2e1efb0373f0dc94fa699c3dfe1e938732bab133","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-09-17 16:11:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27fd820620d264742bd88a7cc31d54b1bfc89fb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 16:56:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/292731cf4c4447b4b35d1d2d83a1fa22 : SUCCESS in 13m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/333dc114d68a4a89a96f0394e29a6228 : SUCCESS in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e61f4bb2f7f461898fce5d1f1bb376c : FAILURE in 13m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b57ce53734441e1ad64c2966189bded : SUCCESS in 9m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff44d35934bc49b2a1682fa9c23f10fe : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d938f1c62444b12b048965be3c9ca33 : SUCCESS in 9m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d351431337f149b4a0f403cd76b9ad97 : SUCCESS in 13m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/00d1d1f2b5d847bf8adce00d5d53f910 : SUCCESS in 9m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4706736d285b40d792a285fba7de905b : SUCCESS in 11m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/656678c33ce94ddc9631e58df98a9f75 : SUCCESS in 8m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fc71a95689ad4ebfbc6bca4e54bcda04 : SUCCESS in 11m 07s\n\nWarning:\n  Comments left for invalid file octavia/common/constants.py\n  Comments left for invalid file octavia/common/utils.py\n  Comments left for invalid file octavia/common/config.py\n  Comments left for invalid file octavia/common/validate.py\n  Comments left for invalid file octavia/common/data_models.py\n  Comments left for invalid file octavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n  Comments left for invalid file octavia/common/jinja/haproxy/split_listeners/jinja_cfg.py\n  Comments left for invalid file octavia/compute/drivers/nova_driver.py\n  Comments left for invalid file octavia/api/common/pagination.py\n  Comments left for invalid file octavia/api/v2/controllers/base.py\n  Comments left for invalid file octavia/api/root_controller.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/api/v2/controllers/health_monitor.py\n  Comments left for invalid file octavia/api/drivers/driver_agent/driver_listener.py\n  Comments left for invalid file octavia/cmd/health_checker.py\n  Comments left for invalid file octavia/cmd/interface.py\n  Comments left for invalid file octavia/cmd/agent.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v1/driver.py\n  Comments left for invalid file octavia/policies/pool.py\n  Comments left for invalid file octavia/cmd/driver_agent.py\n  Comments left for invalid file octavia/policies/availability_zone.py\n  Comments left for invalid file octavia/policies/l7rule.py\n  Comments left for invalid file octavia/policies/flavor_profile.py\n  Comments left for invalid file octavia/policies/healthmonitor.py\n  Comments left for invalid file octavia/policies/provider_flavor.py\n  Comments left for invalid file octavia/policies/availability_zone_profile.py\n  Comments left for invalid file octavia/policies/provider.py\n  Comments left for invalid file octavia/policies/listener.py\n  Comments left for invalid file octavia/policies/amphora.py\n  Comments left for invalid file octavia/policies/provider_availability_zone.py\n  Comments left for invalid file octavia/policies/flavor.py\n  Comments left for invalid file octavia/policies/quota.py\n  Comments left for invalid file octavia/policies/loadbalancer.py\n  Comments left for invalid file octavia/policies/member.py\n  Comments left for invalid file octavia/policies/l7policy.py\n  Comments left for invalid file octavia/api/v2/controllers/load_balancer.py\n  Comments left for invalid file octavia/db/migration/cli.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v2/driver.py\n  Comments left for invalid file octavia/certificates/manager/local.py\n  Comments left for invalid file octavia/network/drivers/neutron/base.py\n  Comments left for invalid file octavia/controller/worker/v2/taskflow_jobboard_driver.py\n  Comments left for invalid file octavia/certificates/generator/local.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/member_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/db/models.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/member_flows.py\n  Comments left for invalid file octavia/controller/queue/v2/consumer.py\n  Comments left for invalid file octavia/amphorae/backends/utils/interface_file.py\n  Comments left for invalid file octavia/amphorae/backends/utils/haproxy_query.py\n  Comments left for invalid file octavia/amphorae/backends/utils/ip_advertisement.py\n  Comments left for invalid file octavia/amphorae/backends/utils/network_namespace.py\n  Comments left for invalid file octavia/amphorae/backends/utils/interface.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/util.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/plug.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/amphora_info.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalived.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/loadbalancer.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalivedlvs.py","accounts_in_message":[],"_revision_number":1},{"id":"b32eb76130f81199f28ee9b048430e60168206bc","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-24 12:00:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"81ffe1732a94e89f1dbdbc47c1f66b80fae60d04","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-24 12:00:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"943988ef83db9566cfffd7de33c5d10579c6f474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 12:26:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63f9b08ef5c342bb878cd077ce81e509 : SUCCESS in 11m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e36f3b8816694980bce86ea2ce090fc9 : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4801769531cf4d389f1f64b53f21c2bd : FAILURE in 11m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92187654ce3e4776a1469d2ad4187582 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de5cfe212a6b40978da9ef6df101da0e : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df82d4fc12684903bc0352542c2207f1 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5faa2ff659154d67a7d946f0a52b029c : SUCCESS in 13m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9b09eb1a01cb49508536068d2b04b6f6 : SUCCESS in 7m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e7ddfc73a01244d18ff5d14af86fb72c : SUCCESS in 11m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2851c9f6386b4a50a9b99641e106fe3d : SUCCESS in 7m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/22039dc9ed014bb09b50288703abc269 : SUCCESS in 10m 35s\n\nWarning:\n  Comments left for invalid file octavia/policies/availability_zone.py\n  Comments left for invalid file octavia/policies/loadbalancer.py\n  Comments left for invalid file octavia/policies/member.py\n  Comments left for invalid file octavia/policies/l7policy.py\n  Comments left for invalid file octavia/policies/quota.py\n  Comments left for invalid file octavia/policies/provider.py\n  Comments left for invalid file octavia/policies/listener.py\n  Comments left for invalid file octavia/policies/flavor.py\n  Comments left for invalid file octavia/policies/healthmonitor.py\n  Comments left for invalid file octavia/policies/l7rule.py\n  Comments left for invalid file octavia/policies/flavor_profile.py\n  Comments left for invalid file octavia/policies/provider_availability_zone.py\n  Comments left for invalid file octavia/policies/provider_flavor.py\n  Comments left for invalid file octavia/policies/availability_zone_profile.py\n  Comments left for invalid file octavia/policies/pool.py\n  Comments left for invalid file octavia/policies/amphora.py\n  Comments left for invalid file octavia/certificates/manager/local.py\n  Comments left for invalid file octavia/certificates/generator/local.py\n  Comments left for invalid file octavia/controller/worker/v2/taskflow_jobboard_driver.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/member_flows.py\n  Comments left for invalid file octavia/controller/queue/v2/consumer.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/member_flows.py\n  Comments left for invalid file octavia/api/drivers/driver_agent/driver_listener.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v2/driver.py\n  Comments left for invalid file octavia/api/root_controller.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v1/driver.py\n  Comments left for invalid file octavia/api/common/pagination.py\n  Comments left for invalid file octavia/api/v2/controllers/load_balancer.py\n  Comments left for invalid file octavia/api/v2/controllers/base.py\n  Comments left for invalid file octavia/cmd/health_checker.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/cmd/interface.py\n  Comments left for invalid file octavia/cmd/driver_agent.py\n  Comments left for invalid file octavia/api/v2/controllers/health_monitor.py\n  Comments left for invalid file octavia/cmd/agent.py\n  Comments left for invalid file octavia/db/migration/cli.py\n  Comments left for invalid file octavia/amphorae/backends/utils/network_namespace.py\n  Comments left for invalid file octavia/amphorae/backends/utils/haproxy_query.py\n  Comments left for invalid file octavia/amphorae/backends/utils/interface_file.py\n  Comments left for invalid file octavia/amphorae/backends/utils/ip_advertisement.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/amphorae/backends/utils/interface.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/amphora_info.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/plug.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/util.py\n  Comments left for invalid file octavia/network/drivers/neutron/base.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalived.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/loadbalancer.py\n  Comments left for invalid file octavia/db/models.py\n  Comments left for invalid file octavia/common/config.py\n  Comments left for invalid file octavia/common/constants.py\n  Comments left for invalid file octavia/common/validate.py\n  Comments left for invalid file octavia/common/data_models.py\n  Comments left for invalid file octavia/common/jinja/haproxy/split_listeners/jinja_cfg.py\n  Comments left for invalid file octavia/common/utils.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalivedlvs.py\n  Comments left for invalid file octavia/compute/drivers/nova_driver.py\n  Comments left for invalid file octavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py","accounts_in_message":[],"_revision_number":1},{"id":"5c29f97be70e8b2bdc25bf4b762f2884f671e682","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-05 15:56:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"43a1d9ef6da8fe3aa7e3903a22cd0e34be166170","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-05 18:35:20.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dfc0e223145b7f59eb2f5b8c028a9b707f84f904","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 20:47:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a97b6bb3478543b9a8899f20f74d970f : SUCCESS in 10m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f018f291d1e493eb19eada27898ed7d : SUCCESS in 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/815d483d3d99478aa3f9ba5719ee8bb0 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d7926c2e07b4227944dbe9890ea5204 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acd5db3258f54c8289924a48f9085742 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2be1b637585e40ebba3480e17f94161f : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47e3108c10bd41ffa0a58d0090f5c862 : SUCCESS in 11m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ad3b1edecc064a5f828cdba0aecc846d : SUCCESS in 5m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/69319ed27d6944a6a80e63a6ab56be03 : SUCCESS in 9m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/256beab3db8249899e5db84390532951 : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/478deeca94f34019a3b153cda7e7b34b : SUCCESS in 9m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c38f3d7eb794ee59cb72ba5cdd83587 : SUCCESS in 2h 06m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/600cb8b3f8c54943a1e3af119e4b000b : SUCCESS in 1h 38m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3b70f8f204af44d48786dca2ab23cd3b : SUCCESS in 38m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3065169f6d584ebea825186378a71b1e : SUCCESS in 55m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4b05c0bbbdd54876b225ecb6dea7074f : SUCCESS in 36m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0e700ed339df4c519701c6db97a7501c : SUCCESS in 1h 27m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dd9bc45f5d7d4774a5c138d733aca587 : SUCCESS in 1h 42m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e29d86e6462e41b79a920446704a231a : SUCCESS in 1h 32m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/607627ba6dc04e929bb55fc179ea89d3 : SUCCESS in 1h 35m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/29cffcebbca64b1881b3ca2ee63662ef : SUCCESS in 1h 20m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a5708624cb464bfe821812c747691774 : FAILURE in 16m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66b5f6f0542b4052028fd5cf82412d379c6819b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 20:48:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"11e9785ac3dbcd8314557efab36a710e96f58292","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 22:40:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a820e5ed69d5481ebb8cc0cabb0dc046 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ad1372b2b1e44ce8c14b82e73665493 : SUCCESS in 8m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ddda4fe8e374ff5a0222dfb623394e1 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08615df524f342e1a7ff041e87b921e6 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cf3906163184eafba8966331c50b33f : SUCCESS in 14m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2e23a4eb72564653980b277b19050c92 : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0ae88e7c14a84b26a16ffa62405c47db : SUCCESS in 8m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fc08272f78b249cba7c1a9e212becd7e : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f23890c6b3854ec6b2843f0a16aa27fa : FAILURE in 1h 50m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/776f19ab437a440e991a7311731a4e00 : SUCCESS in 34m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4d567cf922e44bceaccc3044ef3c0589 : SUCCESS in 1h 12m 42s","accounts_in_message":[],"_revision_number":1},{"id":"39dce848eb92c37eed675d430ab2040fbff7ac81","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-05 22:59:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a65805b59dbceb8391f23f34644836e6131f557c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 01:10:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/951b4a7865314faf8052d7d87e16a03c : SUCCESS in 12m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c376f2257ff24fca88b4373309ad2b26 : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbd4c9ca5498401081258a5b5584fff0 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adaed0d0f4544ffd8ce25cb6a1446c22 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55241254bf754b9a9bbe0227b2d8b06d : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b95e03ce703400195fefd4b602dfaba : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/926652549cda4a68922977b7291afad4 : SUCCESS in 14m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d0c566135c954520b3fd271ef7751201 : SUCCESS in 6m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e79dbccf1aa848d0b50abd6e9a943ebf : SUCCESS in 9m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/52fe539158aa424bb6556afe9600d902 : SUCCESS in 5m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/88cbe62097f34ca89756993d987c6658 : SUCCESS in 8m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee6c3098f85d400f943d1b4510d34726 : SUCCESS in 2h 05m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e38c3ca78d54d1693be6733ea256fda : SUCCESS in 1h 28m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2db65d956d94445f908bf95ea1bdaad2 : SUCCESS in 33m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/558cb337a0294a8eabf5e7df7b9a80a3 : SUCCESS in 48m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c66965eaab9d458181c421692a53e8fd : SUCCESS in 33m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/658b54690df24a8a81bc69a8bce5f944 : SUCCESS in 1h 28m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cab6694ef35c499a89e156d3d80fa50d : SUCCESS in 1h 41m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cba477ebc1024bb5b9963722a31fc634 : FAILURE in 1h 39m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ef2bbe0efc6f463eac6a07c6309ef7b9 : SUCCESS in 1h 19m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/317a1e72964e470daed2fd72b3670ebc : SUCCESS in 1h 24m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/dea43f168eb34b82b4248a38448ff474 : FAILURE in 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5391f70541ce49e7b1b86eb8d8e40951c929b9ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 01:11:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c6ccda108743db31daae478f33442263b148f50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 02:53:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/816717b61aeb482db47b38ff872f484c : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65e358f290eb4bac8d3427aab17e5e40 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f49d26171d914eb98d3fd5e03a89c4bc : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d5da55d0a314fbc8c7a94dc1daecbd2 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa103f7bd7ec4104a8fb1b6ca4a8e406 : SUCCESS in 14m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d717a92b6410460a989a9497e17594a7 : SUCCESS in 5m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4273617da5ee4fc69365bc412af55819 : SUCCESS in 9m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/50922736cafc4e2bbdd2a37d371db7b6 : SUCCESS in 1h 39m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a73e4ccf70cf482598f4382a9db9bb22 : SUCCESS in 1h 24m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5a7c80f2411c4440a490de2e3d908d1d : SUCCESS in 37m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5a22979b715e4478ae3ae86669b30f60 : SUCCESS in 56m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9d9ce8b3763c0d9472fbdfecfd0cd3d979ddf3db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 02:53:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"00915d97e58cde2253ee109b86f1fb07429be64b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 02:55:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91dcb1a6713344dc8839b7b3f6f1712c : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50db4b50b9cd9e40591ab7cacdc5e2555d3b8220","revisions":{"50db4b50b9cd9e40591ab7cacdc5e2555d3b8220":{"kind":"REWORK","_number":1,"created":"2021-09-17 16:11:43.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/01/809801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/809801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/809801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/809801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/809801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/809801/1"}}},"commit":{"parents":[{"commit":"b8ddbc94a1ff1c8dcb3414afd1d5f5269a09f9da","subject":"Merge \"Disable conntrack for TCP flows in the amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8ddbc94a1ff1c8dcb3414afd1d5f5269a09f9da"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-17 16:11:42.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-17 16:11:42.000000000","tz":0},"subject":"Update .gitreview for stable/xena","message":"Update .gitreview for stable/xena\n\nChange-Id: I265b98520fe9708d908079a9dd540b9d6b2de68d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50db4b50b9cd9e40591ab7cacdc5e2555d3b8220"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50db4b50b9cd9e40591ab7cacdc5e2555d3b8220"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
