)]}'
{"id":"openstack%2Foctavia~809802","triplet_id":"openstack%2Foctavia~stable%2Fxena~I5791d335c70eb70af0f5a52190a60bb6da7dc951","project":"openstack/octavia","branch":"stable/xena","topic":"create-xena","hashtags":[],"change_id":"I5791d335c70eb70af0f5a52190a60bb6da7dc951","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","status":"MERGED","created":"2021-09-17 16:11:46.000000000","updated":"2021-10-06 19:17:26.000000000","submitted":"2021-10-06 19:15:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809802-create-xena","meta_rev_id":"cef64f61c734379936dd66536f0e6b09ab54adcf","_number":809802,"virtual_id_number":809802,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-06 19:15:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2021-09-24 12:01:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2021-10-05 15:56:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2021-10-05 18:35:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 16:57:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-24 12:01:20.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-10-05 15:56:32.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"422dafdd4ac0af76bfe23d1e8a91d58b428d5fa3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-09-17 16:11:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbe5e76eddf97d1cb540065bf15595fc167367a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 16:57:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd324190e21b4aab826d84ad16afbc5a : SUCCESS in 15m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a973240c25848d986413d493e9dd84c : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f2d76819683466388bd07cacf483207 : FAILURE in 14m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87ff688df7684d66a34a07439bb4aeac : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62cd23d30905405e8337b31d0ed61fe7 : SUCCESS in 8m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49a7c1aafe34484bb1b2dcee4001fb2a : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6fb74bbc4414ec78eec3b2b0630fd28 : SUCCESS in 14m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06857a5f2acd4ab394c7dfe7ecbdf175 : SUCCESS in 6m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/171b6176765b4973a4cba9e59242f9e7 : SUCCESS in 9m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c60fdc8784ff4fb190c28b1c5272746b : SUCCESS in 11m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b8146cb5fef24902bc4e00475fa58456 : SUCCESS in 8m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/91f2c4fd1cae4f539dfe77954318487a : SUCCESS in 10m 08s\n\nWarning:\n  Comments left for invalid file octavia/common/constants.py\n  Comments left for invalid file octavia/common/utils.py\n  Comments left for invalid file octavia/common/config.py\n  Comments left for invalid file octavia/common/validate.py\n  Comments left for invalid file octavia/common/data_models.py\n  Comments left for invalid file octavia/common/jinja/haproxy/split_listeners/jinja_cfg.py\n  Comments left for invalid file octavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n  Comments left for invalid file octavia/compute/drivers/nova_driver.py\n  Comments left for invalid file octavia/api/v2/controllers/load_balancer.py\n  Comments left for invalid file octavia/api/v2/controllers/base.py\n  Comments left for invalid file octavia/api/root_controller.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/api/drivers/driver_agent/driver_listener.py\n  Comments left for invalid file octavia/api/common/pagination.py\n  Comments left for invalid file octavia/api/v2/controllers/health_monitor.py\n  Comments left for invalid file octavia/cmd/interface.py\n  Comments left for invalid file octavia/cmd/health_checker.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v1/driver.py\n  Comments left for invalid file octavia/policies/pool.py\n  Comments left for invalid file octavia/policies/availability_zone.py\n  Comments left for invalid file octavia/policies/l7rule.py\n  Comments left for invalid file octavia/policies/flavor_profile.py\n  Comments left for invalid file octavia/policies/healthmonitor.py\n  Comments left for invalid file octavia/policies/availability_zone_profile.py\n  Comments left for invalid file octavia/policies/provider_flavor.py\n  Comments left for invalid file octavia/policies/provider.py\n  Comments left for invalid file octavia/policies/listener.py\n  Comments left for invalid file octavia/policies/provider_availability_zone.py\n  Comments left for invalid file octavia/policies/amphora.py\n  Comments left for invalid file octavia/policies/flavor.py\n  Comments left for invalid file octavia/policies/quota.py\n  Comments left for invalid file octavia/policies/loadbalancer.py\n  Comments left for invalid file octavia/policies/member.py\n  Comments left for invalid file octavia/cmd/driver_agent.py\n  Comments left for invalid file octavia/policies/l7policy.py\n  Comments left for invalid file octavia/cmd/agent.py\n  Comments left for invalid file octavia/db/migration/cli.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v2/driver.py\n  Comments left for invalid file octavia/certificates/generator/local.py\n  Comments left for invalid file octavia/certificates/manager/local.py\n  Comments left for invalid file octavia/network/drivers/neutron/base.py\n  Comments left for invalid file octavia/controller/worker/v2/taskflow_jobboard_driver.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/member_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/db/models.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/member_flows.py\n  Comments left for invalid file octavia/controller/queue/v2/consumer.py\n  Comments left for invalid file octavia/amphorae/backends/utils/interface.py\n  Comments left for invalid file octavia/amphorae/backends/utils/interface_file.py\n  Comments left for invalid file octavia/amphorae/backends/utils/haproxy_query.py\n  Comments left for invalid file octavia/amphorae/backends/utils/network_namespace.py\n  Comments left for invalid file octavia/amphorae/backends/utils/ip_advertisement.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/amphora_info.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/plug.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/util.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalivedlvs.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalived.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/loadbalancer.py","accounts_in_message":[],"_revision_number":1},{"id":"5822d8d5c72da0f531874b3c7da78e63b0aa443d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-24 12:01:20.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3096af94a24fe23429003ab79be4841f221635dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 12:38:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5004d353c31d4996a2b1845446c421b6 : SUCCESS in 13m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1423bd3df5d4cb39bef971ecfa84f8e : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f34ce301007448909d127e65b2e36774 : FAILURE in 11m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a01591d9343a42da9f3b3b4dc527ae84 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/641f68242f504cf5a4c6e897ba23c439 : SUCCESS in 7m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af019973d8f34bd1a6adbd1d029a9df2 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52d9ecced6024b329a809e23d2062b9f : SUCCESS in 14m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e3dab78d2444b43957b5c9018d92e50 : SUCCESS in 5m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d57fb6afea3e48d899e1a2fd150207ad : SUCCESS in 6m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/89ebfcbc502e4a5699c22c8966cff691 : SUCCESS in 10m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d3c26f4f5ecc42fcb8a6514582a97b51 : SUCCESS in 6m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9c91210bce054bd8b4d08bdbabbfa11d : SUCCESS in 10m 40s\n\nWarning:\n  Comments left for invalid file octavia/policies/availability_zone.py\n  Comments left for invalid file octavia/policies/loadbalancer.py\n  Comments left for invalid file octavia/policies/member.py\n  Comments left for invalid file octavia/policies/provider.py\n  Comments left for invalid file octavia/policies/l7policy.py\n  Comments left for invalid file octavia/policies/quota.py\n  Comments left for invalid file octavia/policies/listener.py\n  Comments left for invalid file octavia/policies/flavor.py\n  Comments left for invalid file octavia/policies/healthmonitor.py\n  Comments left for invalid file octavia/policies/flavor_profile.py\n  Comments left for invalid file octavia/policies/provider_availability_zone.py\n  Comments left for invalid file octavia/policies/l7rule.py\n  Comments left for invalid file octavia/policies/provider_flavor.py\n  Comments left for invalid file octavia/policies/availability_zone_profile.py\n  Comments left for invalid file octavia/policies/pool.py\n  Comments left for invalid file octavia/policies/amphora.py\n  Comments left for invalid file octavia/certificates/manager/local.py\n  Comments left for invalid file octavia/certificates/generator/local.py\n  Comments left for invalid file octavia/controller/queue/v2/consumer.py\n  Comments left for invalid file octavia/controller/worker/v2/taskflow_jobboard_driver.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/member_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/member_flows.py\n  Comments left for invalid file octavia/api/drivers/driver_agent/driver_listener.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v1/driver.py\n  Comments left for invalid file octavia/api/root_controller.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v2/driver.py\n  Comments left for invalid file octavia/api/common/pagination.py\n  Comments left for invalid file octavia/api/v2/controllers/health_monitor.py\n  Comments left for invalid file octavia/api/v2/controllers/load_balancer.py\n  Comments left for invalid file octavia/cmd/interface.py\n  Comments left for invalid file octavia/cmd/health_checker.py\n  Comments left for invalid file octavia/cmd/driver_agent.py\n  Comments left for invalid file octavia/api/v2/controllers/base.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/cmd/agent.py\n  Comments left for invalid file octavia/db/migration/cli.py\n  Comments left for invalid file octavia/amphorae/backends/utils/network_namespace.py\n  Comments left for invalid file octavia/amphorae/backends/utils/haproxy_query.py\n  Comments left for invalid file octavia/amphorae/backends/utils/interface_file.py\n  Comments left for invalid file octavia/amphorae/backends/utils/ip_advertisement.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/amphorae/backends/utils/interface.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/amphora_info.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/plug.py\n  Comments left for invalid file octavia/network/drivers/neutron/base.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalived.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/db/models.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/loadbalancer.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/util.py\n  Comments left for invalid file octavia/common/constants.py\n  Comments left for invalid file octavia/common/validate.py\n  Comments left for invalid file octavia/common/data_models.py\n  Comments left for invalid file octavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalivedlvs.py\n  Comments left for invalid file octavia/common/jinja/haproxy/split_listeners/jinja_cfg.py\n  Comments left for invalid file octavia/common/utils.py\n  Comments left for invalid file octavia/common/config.py\n  Comments left for invalid file octavia/compute/drivers/nova_driver.py","accounts_in_message":[],"_revision_number":1},{"id":"0f52eb6c02e6079b8bebdd5be0f5942ee214a065","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-05 15:56:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e6666fb5be46a10d6a125e31f8c38666d7ec0fd3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-05 18:35:40.000000000","message":"Patch Set 1: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8a963940c6924fb61070de8b1d4b9ce49815d91e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 20:58:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f45009f5123541aab0921e8b916d9d12 : SUCCESS in 10m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17450a7855204dcf963e75eb938e9bb1 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50c0d3a5291e46608ae7821f250a8df9 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f1f2d18db00443da4b67913c2a30e99 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bd1634c79bf46b3b89d2f6d20aaa3b1 : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc9a5b0348ac436ab801997b50e117a0 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c10235017f584ae6a90d91c176730b73 : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8eff0affeced47b0a260f675d1944e53 : SUCCESS in 6m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ea59f56df67a4ee4a90d55dd13afc52f : SUCCESS in 6m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/34587b5d97b445b9b1b93441ceff6a19 : SUCCESS in 8m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b00a1905b1fb43248e946bcd3b4c9703 : SUCCESS in 6m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eb8b41a9fc4d4b4baf675b0adcb9d32e : SUCCESS in 8m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8c64afa38753436d9afe57b375c5dd07 : SUCCESS in 2h 14m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dc4c014b169d4cf49d3a06eea63a4e9e : SUCCESS in 1h 23m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0bc8d214bef942858ba5d9bd53b96bf8 : SUCCESS in 37m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f76a8f5c6ce942b1a91e286d242db0ca : SUCCESS in 52m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2f92c4ba374e4c29970b55f7f159127d : SUCCESS in 35m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f0a86adbcb85477da9d9cb5f0dbfb361 : SUCCESS in 1h 53m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9a782ecb87ce4392805bc03dbbe1395b : SUCCESS in 1h 38m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b25c0ca7b8cf424cb31d4cc4a8e7c2f1 : SUCCESS in 1h 52m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/94892fbd450f45c7817cb08a62c99c4d : SUCCESS in 2h 00m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0f2301d36f704472a654cee258f226d7 : SUCCESS in 1h 18m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/bbb1e03f140b482bbeac15af54a771e6 : FAILURE in 10m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"46d31bd1b8c15902da86b0747c7756d1bde0eb8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 20:58:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"365696a76737d292319ea7d1d4acbdda6ee2b91b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 22:40:35.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"507e87154fd09db9c70f59e47574266a2a145f38","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-05 22:59:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"367090a278014a3ef621b46142d5927f92b4912b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 01:14:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76f7cf8124f24fffb0bbbd0e7784f7f0 : SUCCESS in 10m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef2f75048be74b91a21044a69fdee079 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b25ed342e9bb448d98c3853e2da66f3c : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a691e54c636a448bac0aaa50a98425f3 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa6c9fd382a24c0cbab8518a16aa9ad5 : SUCCESS in 9m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/678971fb706049ddb2cf6c608b480c5f : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fee20c2a3e0844b9b8106cf93634919f : SUCCESS in 16m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01a24aaf56a24542acb4cf8865730563 : SUCCESS in 6m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9df254140e79465780e1a36c0439528e : SUCCESS in 8m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ccd024337ab744809404fd96c4c9a5a7 : SUCCESS in 11m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/849d48d5ffa84e1b8e539b5ad5ad3e2f : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1f06ea3f01fb44b392ffa67bf050f41b : SUCCESS in 8m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/26c19a809c0b429ea7d8aa269b2aa1dc : SUCCESS in 2h 10m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/908bbeea18a5430dbb2fdc7400c1fe8e : SUCCESS in 1h 27m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/59490b58638e464b9abdc3a36f90b21d : SUCCESS in 33m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/87355cee118748baa04ebfb1dab4ec07 : SUCCESS in 47m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab2366de7a0844398ba06b831429aa0e : SUCCESS in 34m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b70dd15e59e84d25acd7dcc6d1fd312b : FAILURE in 1h 24m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5d4a376672b84f919b0b380a86518314 : SUCCESS in 1h 53m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/db5dc7c038b140fdb86bbe1c8b71f2c7 : SUCCESS in 1h 36m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a573944ed0e44379a8d76f16e8c11443 : FAILURE in 1h 27m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/732ae8c170f645a8ac43fedd64c1cff7 : SUCCESS in 1h 33m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fa3500e1d3db4930b13f601715ac09bf : FAILURE in 13m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fff175f3263b97d658e01205434d18f5d902ae52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 01:15:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12dbe92467eac94c39dbe80e48e86136d0007055","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 03:01:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd3a3a9de48441c4870e7f88cb8d9a15 : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/940beb7072544810b43eb3db7357d74b : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5372f5dba61e4f228913c7f7407beaf8 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d5d34aa8c7f4a90967a00892de4e930 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9d4adab7f3e4c2c8d4b1fc34af17f85 : SUCCESS in 15m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85f0eebc9ca946099f92bd3a561e76d5 : SUCCESS in 6m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9a145872c6e24d5d81335dc7e6e2a7a4 : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/47897f999e3d41de909c070c04440c08 : SUCCESS in 10m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3bd928e214684289b0f6f5876861e4b7 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a2743c5c8dc4c799a1c5fcd10976f3a : FAILURE in 1h 40m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0512deb18e94411d90f252cbca2ed701 : SUCCESS in 35m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/01472bc3272b4c7fab5cc36cdbe67e44 : SUCCESS in 1h 11m 31s","accounts_in_message":[],"_revision_number":1},{"id":"8a835c0181f60def8ad17e8de36aa869f98a644b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-06 06:19:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e5e234fe87ecd990a89ece208fbdc12201277504","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 08:52:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4451db1a95c471ab9d64bd7e8881834 : SUCCESS in 11m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25001af1bc0a440aa29e5658b9435a53 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5b6bb5d4f7044d5a516e102d1334d91 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/493fc32a1c0341428ffd2b85e3ccb562 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f21db6176fdc4ea2aa5457fd105ddb1c : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b47f9016bdbf4c0a8352d2afe6602beb : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6b139fe78ef4a65967662e831ea7189 : SUCCESS in 13m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5b61b654f024edd93be7e1e1dc6eb54 : SUCCESS in 6m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/96eeb98b98964186b4fdae10a9ce838c : SUCCESS in 7m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/04c0dbbf0d184207b685ad20916d50ef : SUCCESS in 9m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5f95f09c04ff410f956ce9b4368224a4 : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5274183a90e548b0aabe1f6b6d210cf5 : SUCCESS in 8m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2965dda28fba42c7869e5232084f7dfe : SUCCESS in 2h 09m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b73a58e4b79f4253b8c9c07fc08cd0b7 : SUCCESS in 1h 34m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/57f2c063b0bb4fd38d1a42296c97cedd : SUCCESS in 39m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4fb0c2f68b1244f79393a6ef6c6ca823 : SUCCESS in 1h 11m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/24ec54883ba04160a91c2b1fd8c29b99 : SUCCESS in 37m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c64f18339a8045c4b11f8d3e9e7f2ee9 : SUCCESS in 1h 28m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3ee7173e92144c2cb821297f46bea174 : SUCCESS in 1h 57m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/33dc93eefb7d488c887ae21fa50f6272 : FAILURE in 23m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ec46c89001df429c90b3c23b5c36e721 : FAILURE in 1h 30m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3b297ffc03244da59eb391aaaa1d31b3 : SUCCESS in 1h 42m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5cd2ea88b75a4f00bdacd76df3933811 : FAILURE in 12m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d866fc9267e3b74af522e802e662b3cc6318414","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 08:54:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23913ac6227998e44fdfa29f211ace2b16916fda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 10:26:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9647185f08943f7b38ea7c2caccc1b1 : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae18a403e5cb4720bdfd03f514cd785e : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8af02f87d124fe4a743eefe5ca7f187 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a50eb13d2939419ca1ea06d01a10be32 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3967e84a5c2e4df1bb6c67a695a439c5 : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43b97e8b53f747718b8f86c8e19d65ab : SUCCESS in 6m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c4738d5751194becaf59617610472a48 : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7d5c4f3833d3418bb8ade87d62ed9ec8 : SUCCESS in 9m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/be0aafcf9d3c403394430a18e3215d37 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0d02c4d5108e4d999ce8739794f2a5ca : FAILURE in 1h 29m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d6ce116262c94f97a0184982cbf32bb4 : SUCCESS in 1h 04m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b19d762577bb4704b3532a306e78c809 : SUCCESS in 1h 28m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9f04613a45cb1a86b27e3a807c13f66a99c20e2c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-06 13:51:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f4da68f006dcf691a04bd91b5014d4f057b84649","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 16:54:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bf0c61dccc74e48a8ed6826de2984cd : SUCCESS in 11m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/233147bf2ce844f09d628b21ddb3327c : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/459246731f844b9896d344844145f432 : SUCCESS in 8m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b31b81cfd311416188fb75d7c488d8bd : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93e9c95898a74eb29f604b783be19b14 : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8535e19961b94de4b0eb595d614d5cd9 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4440eca437e749629c09f4e8152a3f4f : SUCCESS in 16m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1ece23820784a49b33c82370591da1c : SUCCESS in 6m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fe09ed5167654f05a830e91e2ab510a6 : SUCCESS in 8m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/891b8443d16c42d1a212004cee7c97ec : SUCCESS in 13m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a52e36d17ee94201a3808d4b4e572b15 : SUCCESS in 7m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/40815c01d8fb4ab68262fa28c50d5b93 : SUCCESS in 9m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8d00e6815e50400dad57c1f55b569751 : SUCCESS in 2h 25m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e5f21b221bfd4349b45c4fd9018c8d2a : SUCCESS in 1h 36m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6ce05606f9b54b2498c037cd13d0c673 : SUCCESS in 38m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/05b332aeb9e94a578dbc30a3dc91228f : SUCCESS in 1h 24m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6885ac43f853493693389b3290fbe586 : SUCCESS in 46m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c39d82d36c8943568c4ffd50e60cabb0 : SUCCESS in 1h 22m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8e6a4ca7fcee41808514cebf7ce28ae6 : SUCCESS in 2h 58m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4d353f67ce8e4f8baea9b8ec9d9737cc : SUCCESS in 1h 38m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c9bb990090354e438d32f475b790625a : SUCCESS in 2h 03m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8476e75b0e4842a1b99079064aa1c122 : SUCCESS in 1h 19m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1119c4a84bbe4742a061db7df7de3ddf : FAILURE in 12m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"61a80fdec589797c5a170ce0d41605bb11a2a99e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 16:55:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58343dbf850c7b8536a05ba4b3a5d085c2181035","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 19:15:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6757c9d93b944693878a4e813dfe5bc0 : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/938ee08021134e36b21b58813361bae7 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c50e02c69f8249b99a71977bcfcd6b29 : SUCCESS in 8m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c8c8f5aab0b4cd78df95ae2a17b5d72 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdbf7b69aec347d3a63ca20b193f6766 : SUCCESS in 15m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45c93bfa077b442097c420d7190fc719 : SUCCESS in 6m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fbd860f4c8c646e99ea78bf09ac89db3 : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/68c9c61d5c62415a86387288d07fe413 : SUCCESS in 11m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1a7ac82e993c4067a372ec367cc44480 : SUCCESS in 2h 14m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1665d6c2d36947a6ac03b2e6bb76d035 : SUCCESS in 2h 07m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/845530381e9c49a48feb0b37dc581cf8 : SUCCESS in 1h 00m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3ea5ea7eaa264505bf727113d77a8e81 : SUCCESS in 1h 10m 56s","accounts_in_message":[],"_revision_number":1},{"id":"dbf11d4915ad1bbba6628d5d6dacdb6471f7ad9f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 19:15:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cef64f61c734379936dd66536f0e6b09ab54adcf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 19:17:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/815fd4cfe10245768077ae3c98d1325c : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00a3e3820b6f415e97a91c18d7d6c3aa : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a741bcea4f114af9e84a84870a9bb37df5557ba8","revisions":{"a741bcea4f114af9e84a84870a9bb37df5557ba8":{"kind":"REWORK","_number":1,"created":"2021-09-17 16:11:46.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/02/809802/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/809802/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/809802/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/809802/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/809802/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/809802/1"}}},"commit":{"parents":[{"commit":"50db4b50b9cd9e40591ab7cacdc5e2555d3b8220","subject":"Update .gitreview for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50db4b50b9cd9e40591ab7cacdc5e2555d3b8220"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-17 16:11:44.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-17 16:11:44.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","message":"Update TOX_CONSTRAINTS_FILE for stable/xena\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/xena branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I5791d335c70eb70af0f5a52190a60bb6da7dc951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a741bcea4f114af9e84a84870a9bb37df5557ba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a741bcea4f114af9e84a84870a9bb37df5557ba8"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
