)]}'
{"id":"openstack%2Foctavia~809996","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~Ia15a110c6c5a9b83a29bab56538fa425d994d477","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2021-12-02 02:01:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia15a110c6c5a9b83a29bab56538fa425d994d477","subject":"Disable conntrack for TCP flows in the amphora","status":"MERGED","created":"2021-09-20 10:12:53.000000000","updated":"2021-12-02 02:01:47.000000000","submitted":"2021-12-02 02:01:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809996","meta_rev_id":"94a5f0af7b780f3d29764fbc54248df2002fb269","_number":809996,"virtual_id_number":809996,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-02 02:01:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-09-20 14:11:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-12-01 22:00:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-12-01 22:00:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-20 12:40:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-20 14:11:48.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-12-01 22:00:51.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"5c0bfe7d3e2c63e9c784cf4e712f8079a5515e4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-20 10:12:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a61888d4eadc16f2bfbcf350246b95968bb74664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 12:40:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d17f66e423384d509c0a09ab6b8056ab : SUCCESS in 11m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80cd5fac5c3e49239012c2ad91a36aa2 : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bc50b2ae09a4be486371a7838ca4b7d : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/add2a81d4a5b428ca4dfbcd925b6549e : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c37a9252d88044c28d87fe23f7e7a92a : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1740e3b451142eb95a4b9bde3800c48 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b08257ef917a482f860d5a85752465e9 : SUCCESS in 12m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2110534c73f84e5195bbc6eb7f7ce42d : SUCCESS in 5m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2f29df2c676249e5be845f302ee4ab0a : SUCCESS in 7m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b022811c3bdf4844883af547dcf22a17 : SUCCESS in 9m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/88ed3c72f728467aa03337d011d3b8a2 : SUCCESS in 7m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9a04ca1d5b354618b3c709352485f150 : SUCCESS in 9m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/835410909e1f4a5a924586b56d08de34 : SUCCESS in 1h 32m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/70260ef6cb2443f4843bb2f44efc87b4 : SUCCESS in 2h 25m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/44cb3c42161849ffa083e42d897c5023 : SUCCESS in 48m 23s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4e4b400f20894ed19ca6715b08a347d3 : SUCCESS in 45m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a907436e29e44aefbd89314a38ca088b : SUCCESS in 32m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7d655d7e453b43fda8bb674cfad7ee89 : SUCCESS in 52m 44s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bd3a81961472493091285423ca558e78 : SUCCESS in 2h 02m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/42b2cfc377264eda85ca56a57278990b : FAILURE in 33m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a1c22b301bad45d3aae8c8d86cd1cc3b : SUCCESS in 1h 48m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f8a3de31561d4c0c944c1d38b938ac6e : SUCCESS in 1h 38m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/47d8d610985640f2b70fb1b665bd86ba : SUCCESS in 1h 27m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/89eb26eeb9904dc88ae4e298dc2b50ea : FAILURE in 11m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7c52b7b87b8bed047830c3dcb2d87c7b86f9e85d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-09-20 14:11:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e2ea5f517c2adc6857850e87f8f5e44b7bb9146b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-12-01 22:00:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fca23dad4674e46e13bbb7b7ca38acbc630d6c5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 22:01:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f4c8217f1894be4503f9748b5a858eded4271d76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 23:48:37.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea6865c7da91d4b0f7e7dc54d07f1c642db809e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 02:01:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/313154ba380240ce8cd5d9beed948086 : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81d746b0fc0a4c1b81c803f78e7533c0 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/015356cc2e02449294632ab167022d59 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ee1ed4fd8f34c6086cfbeccd33722d7 : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d0b8520e65940a2ae39302dab80a305 : SUCCESS in 14m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7742712ec89c4deda9a56719e29c5b57 : SUCCESS in 7m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/64c08117caa149dba651bd725b53ade5 : SUCCESS in 7m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6f933919f19a4dd18be23489e00c6fa6 : SUCCESS in 8m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2d593238091a4a20ab20f9975c4b4bfb : SUCCESS in 1h 37m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f818aa9337384e0781b121c5ce1bd403 : SUCCESS in 1h 13m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/233b436213e14b53b90134d3f30d83e2 : SUCCESS in 39m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/831299107ce345daa552122aff29e92a : SUCCESS in 31m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc8cd8b7e8bc47cba9aaca61b5b942d9 : SUCCESS in 32m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bcbdd0e0b40a46f7b2bcd60b7e0bcca4 : SUCCESS in 47m 48s","accounts_in_message":[],"_revision_number":1},{"id":"94a5f0af7b780f3d29764fbc54248df2002fb269","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 02:01:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1905859662a75403705db2fca6619b929814fc6d","revisions":{"1905859662a75403705db2fca6619b929814fc6d":{"kind":"REWORK","_number":1,"created":"2021-09-20 10:12:53.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/96/809996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/809996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/809996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/809996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/809996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/809996/1"}}},"commit":{"parents":[{"commit":"f16f72cca2a1fec54896cad72100607f2acc9b7f","subject":"Add caps for pip-extra-reqs/pip-missing-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f16f72cca2a1fec54896cad72100607f2acc9b7f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-15 07:00:52.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-20 10:09:03.000000000","tz":120},"subject":"Disable conntrack for TCP flows in the amphora","message":"Disable conntrack for TCP flows in the amphora\n\nconntrack is used in the amphora for LVS-based listeners (UDP and\nSCTP) and it\u0027s enabled by default.\nAdd nft/iptable rules to disable conntrack for TCP flows, it reduces\nmemory usage and prevents getting \"nf_conntrack: table full, dropping\npacket\" messages.\n\nStory: 2008979\nTask: 42623\n\nChange-Id: Ia15a110c6c5a9b83a29bab56538fa425d994d477\n(cherry picked from commit b3f59eeb73289306dd5fd2e9d664903911dc614c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1905859662a75403705db2fca6619b929814fc6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1905859662a75403705db2fca6619b929814fc6d"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
