)]}'
{"id":"openstack%2Foctavia~809997","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~Ia15a110c6c5a9b83a29bab56538fa425d994d477","project":"openstack/octavia","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2021-12-02 02:02:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia15a110c6c5a9b83a29bab56538fa425d994d477","subject":"Disable conntrack for TCP flows in the amphora","status":"MERGED","created":"2021-09-20 10:16:28.000000000","updated":"2021-12-02 02:02:09.000000000","submitted":"2021-12-02 02:02:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809997","meta_rev_id":"46490022516550b18b3b60af5ae8640b8f65b307","_number":809997,"virtual_id_number":809997,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-02 02:02:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-09-20 14:11:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-10-27 12:52:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2021-12-01 22:01:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2021-12-01 22:01:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-20 12:26:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-20 14:11:55.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-10-27 12:52:27.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2021-12-01 22:01:58.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"249a937525f0bd06811a80d08398ebd185117519","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-20 10:16:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afefb23d8434a15f537950a663d755f4e5d60404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 12:26:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00a888ac21774552b1c78e9eeebf6df4 : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4649e20a8324bc59cc06f6bfae2bea0 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57088c6beae7400386300a7b8b094ea7 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2495f446926b49ab9eeecd73a6d4a5ab : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3766b919e27a4467bcd6d6cf48ee88b6 : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56ce1c109bf241ddaf205950dead6fee : SUCCESS in 6m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1910af2a857a4dc3911f0d2ad301869b : SUCCESS in 7m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9993bd4a5f934a5a8803919d1024980f : SUCCESS in 9m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/28581bec891f440bb498df18c1c5da20 : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b8056b8900c449b89b0d9a1dbb529b33 : SUCCESS in 7m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/91a62090713a407a8fd5cb8b97aefe04 : SUCCESS in 1h 23m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fbf7ee5b9d4747819615830ac3792c23 : SUCCESS in 1h 18m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3631b5fdbc4e4f1d9bb253b0e46a7cb5 : SUCCESS in 35m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/39d640bac9fa4ffba531f82abf40d061 : SUCCESS in 35m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dbd61a24a63a447cb497cb9d67062a7c : SUCCESS in 31m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/606db508e56447da97738c5ed7d5b1db : SUCCESS in 1h 00m 59s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/874932c61ca94c20be507c1a0b5d3373 : SUCCESS in 1h 30m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a9852f65ba3c45288132b627109c90bc : FAILURE in 39m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cf9bbda0c58c426cb683fd74459f5e59 : SUCCESS in 1h 25m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e3a81fc70d134a2db722cac8a79d0658 : SUCCESS in 2h 08m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1ec430ac26134ec8afc66cb264d95a36 : FAILURE in 1h 27m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fac1aba6c36642d188f7e32ac7910db4 : FAILURE in 30m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"09209d839c929a468d226b8562fe2292c63ed286","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-09-20 14:11:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"78554e7589b0c5e67ce6da782005a18e0287886e","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2021-10-27 12:52:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"72e3d650cab67f38ddcf833bf34859de41227385","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-12-01 22:01:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"16ca6073c7237a77166db148e0d75aba311bbdc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 22:03:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc506f23a4b781fb4a44e741d688196862cfc50f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 23:49:30.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47e4a39b7cde0a94b2eb3779a8b788b145a446ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 02:02:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df4730db7e0f4119b9471b40b68c0363 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2de862cf29774a28b37ccdae65f314cc : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f683e456e14f47b9823d43cdfe3f58c1 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8e180c881cb4019a9d4a6c3170c2648 : SUCCESS in 12m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2324d6dc697a48d0ba3634a1171acc66 : SUCCESS in 5m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0aae5ce3ef89441885973590007bb269 : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7317179751e349c3a6a84ba2875c109d : SUCCESS in 8m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/154d049d142b44318e46575f0ec340dc : SUCCESS in 1h 29m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/66167b1aaee1413d946e8478021e5de3 : SUCCESS in 1h 18m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/15b1918146364c19a324cdf5fd6fd06a : SUCCESS in 35m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/970962081dd24b4398045b73ed13d8a5 : SUCCESS in 32m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/795e05afa0fb491284253d325da0c15c : SUCCESS in 33m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f07cfc52f05a481992823936e6d9aec1 : SUCCESS in 48m 35s","accounts_in_message":[],"_revision_number":1},{"id":"46490022516550b18b3b60af5ae8640b8f65b307","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 02:02:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c0ad3b347d514d6851c1c9234af0dc2d5c2a6952","revisions":{"c0ad3b347d514d6851c1c9234af0dc2d5c2a6952":{"kind":"REWORK","_number":1,"created":"2021-09-20 10:16:28.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/97/809997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/97/809997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/809997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/809997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/809997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/97/809997/1"}}},"commit":{"parents":[{"commit":"f3b3e35af34667e2a7d718cc203f2ee03105c29e","subject":"Merge \"Fix race conditions between API and worker DB calls\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f3b3e35af34667e2a7d718cc203f2ee03105c29e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-06-15 07:00:52.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-09-20 10:15:34.000000000","tz":120},"subject":"Disable conntrack for TCP flows in the amphora","message":"Disable conntrack for TCP flows in the amphora\n\nconntrack is used in the amphora for LVS-based listeners (UDP and\nSCTP) and it\u0027s enabled by default.\nAdd nft/iptable rules to disable conntrack for TCP flows, it reduces\nmemory usage and prevents getting \"nf_conntrack: table full, dropping\npacket\" messages.\n\nStory: 2008979\nTask: 42623\n\nChange-Id: Ia15a110c6c5a9b83a29bab56538fa425d994d477\n(cherry picked from commit b3f59eeb73289306dd5fd2e9d664903911dc614c)\n(cherry picked from commit 1905859662a75403705db2fca6619b929814fc6d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c0ad3b347d514d6851c1c9234af0dc2d5c2a6952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c0ad3b347d514d6851c1c9234af0dc2d5c2a6952"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
